)]}'
{"id":"openstack%2Fkayobe~736485","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I2dbaed7d698b630b1c79cf72fa338e29148dd6e7","project":"openstack/kayobe","branch":"stable/ussuri","topic":"openrc-upgrade-stable/ussuri","hashtags":[],"change_id":"I2dbaed7d698b630b1c79cf72fa338e29148dd6e7","subject":"Update openrc files after overcloud service upgrade","status":"MERGED","created":"2020-06-18 07:54:30.000000000","updated":"2020-06-19 11:09:40.000000000","submitted":"2020-06-19 11:07:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736485-1592564862629-09deae80","meta_rev_id":"558819394a3446f9a6ad6c78650f2dcde79e4258","_number":736485,"virtual_id_number":736485,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 11:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 10:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 21:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 21:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 21:23:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 11:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96cf78558c8ec6d3167a78b0e53bd3ec9e364bae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:54:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"98fabc8749fb1f83d785d4dc2d41f563e9c93317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:30:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6866d2311993475c80aaa7e8ff0b9e7b : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98b4333b6954b8cb2f2f27d16c2ca94 : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6657fadebc4745a9b43b5c5c18e997e0 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c28573bc55fb449f955375645da01965 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4534e995a432439ba2eb7942d3ad9269 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d573e66764904ae88e60f952c4b4016c : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c100bbca7bf4d07be29b226ef650f74 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6d3cb61676b466d8ab60e4783d22460 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a6996d6fa484847ade2a10cd9f1d800 : SUCCESS in 35m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/58dcd567e5174dbc9e678cae6d25f278 : SUCCESS in 1h 22m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cda5eac1c3074068bd81d8d30bb68bb7 : SUCCESS in 18m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/851aeb9155ed4cad80af9691fcf6e92f : SUCCESS in 2h 03m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ca935ad37b3d4de88bbf0d3f3861b62e : SUCCESS in 18m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4de4c498872949608dd5baad1212099d : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":1},{"id":"26798e341f45460dbd8d62e9bddaf1eddeaf69bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:42:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aaa1e4aa355e24e6aaa39b946fb8c75bf1bd5b4f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 21:23:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"031c02056f0ce1d90cf3d6323620c329dd8240c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c9e7b54402c93a8447382ba6415b45cb1d93f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:20:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b758065a5da64baaa689a03b249eb490 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/896dfd7914d5403091ecf187a8144132 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c707000d48a463cadac13c71596ac78 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4b09cd86d3412f801733128d4a290c : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67d10a14b1c44d49b63443720c8a7e82 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6a662d6ff07479ba4b4bb0fefa31736 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cea4a9d5dcd54fd09daa9e66a3e94e7f : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f72cb4b6a774508beec10922d87e778 : SUCCESS in 34m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1ca6cfba0139452dabdf6549ca874957 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e51f1222bb06407e80472c47693d8d3e : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/da2169074a26459b97989a8edb67ac0d : FAILURE in 1h 50m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/98d77d7858314077aa524564386dff2e : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/13bb3f014fd84ceb9315c5d3b42e959c : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3d2c4c8d15ca3e9dcde1b49d397f9ac6e077e039","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:23:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc66e8d8e0103502a965301373d2bcb40fdd5dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:39:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d61098e4f6d465b9e21834cb74fbc31 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7deb942b611245e2b2caa22614880d9a : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a8621cd8e0e48cdaa9b2003341f2eeb : SUCCESS in 9m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d85f014f963496a9fdc99165470a680 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be2e1d64f3e4d0ebb9ab425ba4ac9d2 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d277b9fe0f174629b5981eced5168692 : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/494406c9082e4dd193fa40e715735280 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ee22833c84f4f05aee67ca7b7942d62 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6156da78152b4c5fbe0cb71822684898 : SUCCESS in 34m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c8d92c470eda465bbf10ff88be0638c1 : SUCCESS in 1h 29m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4b4e4f993eac47eb90aa576111000c9b : SUCCESS in 21m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f8ab271ef880472cba02e704269fe0c7 : SUCCESS in 2h 10m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f61e6f4c4d8452cbb904f08cc373eab : SUCCESS in 25m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/62db0793209044b59472f257cca0771d : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"40cf262d5cfbd06a5220b74dfd3d3a85cb3024af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:40:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"083f697ac43921322fd061b44f41a19643d73cbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:07:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6bd62542af423883c8cce1895a9151 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7380a656b5684bb4a38643f6b86fc611 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f1b8cbc7024789a93875c9dde8fd79 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1722e536d5fa4c6c9239fd7e204a4909 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/969713f51edb4b56af60b100d1b6742f : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d7f235373a4417fa26ff8399a095ecd : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a66d3e1e1b24ac28b2e5e992a694867 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b817baf1beb4748b5dcccf7298fb238 : SUCCESS in 37m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1c51abba2f6345ff9a4486848ef25412 : SUCCESS in 52m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/37fe3b5bcec64cf6bc68671def5f7a4b : SUCCESS in 16m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fc992c8c2dbd4b7b9ca3cc009afb30b6 : SUCCESS in 1h 25m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3a1b77cc01de448d8663851726d3099d : SUCCESS in 21m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d1663c6e17e42f68618a0b2393df6c1 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a117455e2eeedc0a38a5ff4202ff11751125b907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:07:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"27b0f17cdf9e55a33d6058f9412977205d46e6ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 11:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f558ae8c764483bbbfd34ded89f289d : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8233beb207f46e0afb5b3ff4f5f903d : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90e49fdb844059893d80ba3982b9da053f6ec0d4","revisions":{"90e49fdb844059893d80ba3982b9da053f6ec0d4":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:54:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/736485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/736485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/736485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/736485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/736485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/736485/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-05 10:58:15.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:54:30.000000000","tz":0},"subject":"Update openrc files after overcloud service upgrade","message":"Update openrc files after overcloud service upgrade\n\nWe could end up with openrc files not matching the latest templates\nafter an upgrade.\n\nChange-Id: I2dbaed7d698b630b1c79cf72fa338e29148dd6e7\n(cherry picked from commit a633536bfaa8909f453c26fa4f1be2bdb6792fb7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90e49fdb844059893d80ba3982b9da053f6ec0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90e49fdb844059893d80ba3982b9da053f6ec0d4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
