)]}'
{"id":"openstack%2Fkayobe~736486","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I646930ad8ea70991d6ffa00f15f93f72d922141b","project":"openstack/kayobe","branch":"stable/ussuri","topic":"story/2007679-stable/ussuri","hashtags":[],"change_id":"I646930ad8ea70991d6ffa00f15f93f72d922141b","subject":"Copy custom certificates","status":"MERGED","created":"2020-06-18 07:57:07.000000000","updated":"2020-06-26 13:40:08.000000000","submitted":"2020-06-18 22:27:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736486-1592519249775-c5316477","meta_rev_id":"323eb62f39317a85e4364e381172b140f9993255","_number":736486,"virtual_id_number":736486,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 22:27:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 11:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 20:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 20:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-18 07:57:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 20:30:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 22:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4d434d422ef2fe1e180fac7437777627f00403a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:57:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"397e0b978a13da860498bf692f0df9d052b1deb9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:57:47.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ce30894faa0c42d64b87fac945b6051d49a65757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/198fe7c751404fa7866ae9cb00266a80 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc00cde656b4bdeb5b2761b60d536c9 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d645f3311e54f228402050c50d41e69 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6151c8ceed7d4eeb88658c60e1cfb665 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420c3858ca0c42ae828629be6bd301c2 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f948955a6ac945d294010e49bf8ec107 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/266ea38a86044fe4b1540dcb317ca948 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ee7a560b81342d0a29c1bef9c518a1c : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ee7881c0d2049cdb08b5ec5f60de98d : SUCCESS in 36m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e01d31efc7ba4643b08f19956bc1931e : SUCCESS in 1h 36m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e3fa64bba7824f3f8d411999f23ead47 : SUCCESS in 18m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0ca6e7bd2e0940fbbff93326d2eac3d1 : SUCCESS in 2h 25m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d6ebd5b4fc941069e91b026e576d86a : SUCCESS in 18m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/811e3ded3bf04b76bfe22da13818057b : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9aafcc7e2809e2509ff61455191595ac96245992","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 11:14:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48e364ddd3ae38afa4c3b4c2045e810be9cd4c1b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 20:30:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c42f042bf51f395e8f8a065e82ea6dc8ac1c5930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2b34438064e686389d5bbfe267a8b2d767882d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:27:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f63d6fa4a66f0a6b65844983d89aff3abc140399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:27:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf60a80da36465187e3c78118630ff9 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b8b652655948c7a1736e87209b7794 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4358ee021c414b15b9b1dc386f7ffabe : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d85d0994eda4a21b6fb5ec088f30c75 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/751770bde8534dfc8a54318c2209a37e : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b8be3585d7a4181a725b5fd997c3f3b : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2983b7c66887477da893364874c0f1a5 : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a09dc4e41b434e1f9007d105fbec7b9c : SUCCESS in 27m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8744446b8ebd43329d5c5a52312fbef1 : SUCCESS in 1h 30m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/19fbe201a024420c8ee4dea19093af7b : SUCCESS in 17m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/77292f12ac064712a283698964d8694f : SUCCESS in 1h 51m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/524b4b02853b4739a1d0e6028203607d : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/49d37c52c5054c41906829b2407a78c2 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d34c2d4dd45cbb5cd2e93fdb3db73b994010cfc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:30:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433188e1f2da4a17b7a635ee85d536cb : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2426abacc3cc445390beb9dd726853f2 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1},{"id":"224ebb41cd4cce6a3f50727cb09d54d7a0b69494","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-06-26 13:40:08.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 882f0ac49a1a2350929d5c612907218898d6849c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92abb8c007c0bddaf77fe1a304829a5c198fad6a","revisions":{"92abb8c007c0bddaf77fe1a304829a5c198fad6a":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:57:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/736486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/736486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/736486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/736486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/736486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/736486/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-15 10:21:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:57:07.000000000","tz":0},"subject":"Copy custom certificates","message":"Copy custom certificates\n\nVarious kolla-ansible TLS features (including backend TLS and custom CA\ncerts) require certificates to be passed via\n$KOLLA_CONFIG_PATH/certificates/. Currently Kayobe does not support\nthis.\n\nThis change adds support for copying across files from\n$KAYOBE_CONFIG_PATH/kolla/certificates.\n\nIt also uses the kolla-ansible default value for\nkolla_external_fqdn_cert and kolla_internal_fqdn_cert when\nkolla_external_tls_cert and kolla_internal_tls_cert are respectively\nnot set. This allows for the standard kolla-ansible configuration\napproach of dropping these certificates into the\n$KAYOBE_CONFIG_PATH/kolla/certificates directory, rather than defining\nthem as variables. This can be useful if using the kolla-ansible\ncertificates command to generate certificates for testing.\n\nChange-Id: I646930ad8ea70991d6ffa00f15f93f72d922141b\nStory: 2007679\nTask: 39790\n(cherry picked from commit 1d12ca545e33511b7584662d57639ee6a10aa7cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92abb8c007c0bddaf77fe1a304829a5c198fad6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92abb8c007c0bddaf77fe1a304829a5c198fad6a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
