)]}'
{"id":"openstack%2Fkayobe~736716","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ie2a0a6bf75dc88458796a9691891e76f578d87ad","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ie2a0a6bf75dc88458796a9691891e76f578d87ad","subject":"Fix kayobe baremetal compute commands","status":"MERGED","created":"2020-06-18 13:34:42.000000000","updated":"2020-06-18 22:29:16.000000000","submitted":"2020-06-18 22:27:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736716-1592519239182-a63359cf","meta_rev_id":"c2d023812bec7a5a2ce0eb0601621aa6cdf016fe","_number":736716,"virtual_id_number":736716,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 22:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 19:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 20:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 20:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 20:20:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 22:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"234039baffa4bc6e48e4a7a090eccfefcedf18c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 13:34:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c68269d1cdef81e9bf151c5c402ba22f151c1e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:51:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2e45f2410a54bdfb1ab8cf23cb19acd : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5a4935055e4c5899f288ca19a93540 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af3b38b7cf8940a9a3a920ce779cac54 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b3a7d2f43584077ab5b7636f93d240b : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21bf61c137d4278a014f8cb274a4962 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b68b08706aed4243b27865ed3ef256de : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99501e2a8ff844a18a64378544a69729 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b00188a54eb6475d8e88f7d3d1b888bf : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/886a62c9949a4628bd97965f781e6ba6 : SUCCESS in 36m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4b055cb39b51429bb2f74c43d922d115 : SUCCESS in 1h 17m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6fb3e6ed57044fd48f75298b9f2bfe70 : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/15e3af0d60234b378e6352a936bcb28c : FAILURE in 2h 06m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c2a90193940c4b858b9ca81dcc792d7b : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/83d1d233288646cbabe248b3ff3c4542 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c65721deb5780ca31e1361d6df1f9e4c9a36d301","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 16:13:44.000000000","message":"Patch Set 1:\n\nZuul VM\u0027s disk is full, leading to RabbitMQ alarm, leading to blocked connection in nova, leading to the 504 we see when trying to delete the instance.\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/system_logs/df.txt\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/kolla/rabbitmq/rabbit%40centos-8-rax-dfw-0017230756.txt\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/kolla/nova/nova-api.txt\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/ansible/overcloud-test-bm-post-upgrade","accounts_in_message":[],"_revision_number":1},{"id":"9fe5fa2549a880340928af5d439eb3c2da060e86","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 16:13:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e78445516377f77ef03370e9e5a4dd33984da7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db155282cd4d4cddbef08bc2ca6b2fff : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a823fad295624574abe9253cdcb154aa : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c674fb22ff8144b0bfe971048a5505f1 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/629c6bde0c624e37b129e29aebfcb459 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62b1d220e824febbf417a43ed0868d7 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9206e21f18a0413790bf0da6345b4840 : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6283566fa39f4f118dcd6857073fb96d : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bfbe746170f4a34af8f392c509b3071 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/738f84c396f840c79845307739d9a7bf : SUCCESS in 36m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/82b05f196111409c87649f31964cdc0e : SUCCESS in 1h 14m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b8795cbc4a4494eaae60b111ab49a5f : SUCCESS in 20m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c57a3020d7b84dc8ab1df2f78c9bca0b : SUCCESS in 2h 00m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f0733431c42c4dd0958fd1ccd1cd513a : SUCCESS in 21m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/22f8859bb5664953addade055c119886 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bcd48a0600aa0be0cde5f88bcb53e163fae8df96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 19:36:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db71ef97eb91787f599163619e22fbed23d189e2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 20:20:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e Zuul VM\u0027s disk is full, leading to RabbitMQ alarm, leading to\n \u003e blocked connection in nova, leading to the 504 we see when trying\n \u003e to delete the instance.\n \u003e \n \u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/system_logs/df.txt\n \u003e \n \u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/kolla/rabbitmq/rabbit%40centos-8-rax-dfw-0017230756.txt\n \u003e \n \u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/kolla/nova/nova-api.txt\n \u003e \n \u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_15e/736716/1/check/kayobe-overcloud-upgrade-centos8/15e3af0/primary/ansible/overcloud-test-bm-post-upgrade\n\nI didn\u0027t realise we were using that much disk space. Would it help to switch to overlay2?","accounts_in_message":[],"_revision_number":1},{"id":"d3d00d79b2619bf1e92043d406d41ac8966aa953","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:20:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27fd2e83c853c726921c35b3585539de30cb6a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:27:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030b8da2865b4d0d949fb3737d7eb1bf : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb796ba435954114ad5c1138176cc56a : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b9639710caa42d8b281628ab2d4c033 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32884101bb7f48de9106f27e3a030026 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e25128a5c746538d98716b42e506f8 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/174803ee9d5d4e12bdc56723a5e86095 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56fefd50333e4a279f31b7a3a6ab1d7e : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15d39efda0e846d1bba51280b91506fb : SUCCESS in 34m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b0703a25efdc485680d25f6846f3dcbf : SUCCESS in 1h 14m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eefb99f87a224350bd585e2fe0c1efeb : SUCCESS in 18m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fcaa70cfab164978b86853fcce28d7a6 : SUCCESS in 2h 01m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7a6cbd5d36e54c9dacad9f534f323fc6 : SUCCESS in 19m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e89aab09d48c4e22bf1657d51fd965d3 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1be82f894fd269014563a0291059d2fc5f12c60e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:27:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2d023812bec7a5a2ce0eb0601621aa6cdf016fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:29:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef068c1867940838b613e9dc961f428 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c2b36ef869040519cf7eae67bf9d455 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4baf2eaf1feb03efebbe6a36a355388fa23f2dd","revisions":{"c4baf2eaf1feb03efebbe6a36a355388fa23f2dd":{"kind":"REWORK","_number":1,"created":"2020-06-18 13:34:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/736716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/736716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/736716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/736716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/736716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/736716/1"}}},"commit":{"parents":[{"commit":"8253d019bf8531aeea105fd93087670f9fd4c4f0","subject":"Merge \"Update image regexes\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8253d019bf8531aeea105fd93087670f9fd4c4f0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 16:48:49.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 13:34:42.000000000","tz":0},"subject":"Fix kayobe baremetal compute commands","message":"Fix kayobe baremetal compute commands\n\nSince adding support for Python 3, the following commands fail:\n\n* kayobe baremetal compute inspect\n* kayobe baremetal compute manage\n* kayobe baremetal compute provide\n\nThere are two issues:\n\n1. There is a missing /bin from the ansible_python_interpreter path used.\n\n2. The stackhpc.os-ironic-nodes role fails with an undefined variable:\n\n    name \u0027openstack_cloud_from_module\u0027 is not defined\n\nThis is fixed in the stackhpc.os-ironic-nodes role version 1.3.1.\n\nStory: 2007797\nTask: 40051\n\nChange-Id: Ie2a0a6bf75dc88458796a9691891e76f578d87ad\n(cherry picked from commit 08c5d00a619edb79d36656190a47bf4652fb043b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4baf2eaf1feb03efebbe6a36a355388fa23f2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4baf2eaf1feb03efebbe6a36a355388fa23f2dd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
