)]}'
{"id":"openstack%2Fkayobe~736971","triplet_id":"openstack%2Fkayobe~master~I3f96c877b24d4a400578dc618a6dc488c0fdeeef","project":"openstack/kayobe","branch":"master","topic":"bifrost-kernel-options","hashtags":[],"change_id":"I3f96c877b24d4a400578dc618a6dc488c0fdeeef","subject":"Support configuration of Bifrost\u0027s Ironic extra kernel parameters","status":"MERGED","created":"2020-06-19 10:19:43.000000000","updated":"2020-07-28 17:12:09.000000000","submitted":"2020-07-28 17:10:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736971-1595956200652-4e4c0980","meta_rev_id":"b0bb474f53b50922da9c808f92cb60a8ed1cb259","_number":736971,"virtual_id_number":736971,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 17:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-07-28 14:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-07-28 14:40:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-19 14:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-07-28 14:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 14:20:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-28 14:39:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-07-28 14:40:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-28 17:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d13b1eeb37f395868c5769271adcfb7da5d64520","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 10:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1be1916f25dabbd67ffd1d2640375619a89ca3b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:12:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6b14d1e6894402fba731987f75dd5f7 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3305682197cd41b5ab5e57f522ee5c93 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e25703ea2d04432a6e3fe6a40aa93dd : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f5652982da842bba6d6c7e3523b4219 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae38a34756084b5aa637bd6c6eb23b7d : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a847a5da0c54499eb28fe6fbca2dcc08 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2eedea6c1e294dd798eeead6442441ae : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e28ee045c054e168bbbc5ab53216673 : SUCCESS in 9m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8e07d188a8a4694b9266031b5a0004e : SUCCESS in 28m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17e8a0b0270840c888caaf30205f1ec4 : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/710ef475735f4e41b6918115b175ee0b : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bd8408ffc1eb44319bd1a0ad84e368ad : SUCCESS in 1h 50m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bff4308cdd9245f39169de0e7a47f04c : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1fe712557ac84d2cb8f7a408047a3f4c : SUCCESS in 26m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/495e89ae138b4a04893d617eba333067 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"816f061932aebf0e2aa26c57fa388e80b6751f24","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 14:20:10.000000000","message":"Patch Set 1: Code-Review+2\n\nFine by me, but FYI you can provide arbitrary options to bifrost via etc/kayobe/kolla/config/bifrost/bifrost.yml.","accounts_in_message":[],"_revision_number":1},{"id":"1e783960c5b99bd797cf494831bf3516f807a879","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-28 14:39:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec480d34f6c8f13ec570f4a97e43d8c8e33f99df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:39:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"212ce5eb6357afcf089129f47b496e4928b809c7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-28 14:40:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7cbfc6b03bd15e0cf1f59f5653fe67ed5754573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:10:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fd826754c3489098631621e81dcf61 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba93167dbdaa4bfd9145aea1127ce6bd : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/619e2637357d4796b345a15c24d703bd : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c41b6b9ae0f4d0d82645c64170fb8f5 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fba571d5644444d8d8e4dd2afad622f : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc5e4fc7916a4ecdaf8274d5818f0b36 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ee365353b3e43528b957664707e69a4 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afc3cb7c81ff4b0796fd8e79e4f71fca : SUCCESS in 20m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/726c1f053eb24499b08d75af16388d84 : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6b673b33e1f04ac6b3ad6eb1aaf2478e : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ba999168fa4b4d1990ae3b436cd12657 : SUCCESS in 2h 06m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd8f311b92b947a59560f3e3a82ae4a8 : SUCCESS in 17m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9a64f9a88a1a4d739ceb4af2f3d33cb0 : SUCCESS in 28m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1328b8d5721449a98cc4158c51289211 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"514d8d9835ab5f838d7af86d5119ee0ff4bf525e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:10:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36bb55f3715412cdc2da071f6e22c5a9ee61613c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:12:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c11d78ea4c74bf88065238fd13b7cc7 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a8e636c2bd64470aef6844eb0ea814d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b44571248b6251942928538da5689ecf7b81abe","revisions":{"5b44571248b6251942928538da5689ecf7b81abe":{"kind":"REWORK","_number":1,"created":"2020-06-19 10:19:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/736971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/736971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/736971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/736971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/736971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/736971/1"}}},"commit":{"parents":[{"commit":"da95fc897d9306f2fc082feff5497e9b6a206c05","subject":"Merge \"Fix kayobe baremetal compute commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da95fc897d9306f2fc082feff5497e9b6a206c05"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-19 10:19:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-19 10:19:16.000000000","tz":120},"subject":"Support configuration of Bifrost\u0027s Ironic extra kernel parameters","message":"Support configuration of Bifrost\u0027s Ironic extra kernel parameters\n\nChange-Id: I3f96c877b24d4a400578dc618a6dc488c0fdeeef\nStory: 2001987\nTask: 19607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b44571248b6251942928538da5689ecf7b81abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b44571248b6251942928538da5689ecf7b81abe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
