)]}'
{"id":"openstack%2Fkayobe~736976","triplet_id":"openstack%2Fkayobe~master~Ib27deac4ddf4976a571b192ee5d1b8ca57701916","project":"openstack/kayobe","branch":"master","topic":"custom-route-options","hashtags":[],"change_id":"Ib27deac4ddf4976a571b192ee5d1b8ca57701916","subject":"Add support for custom options in static routes","status":"MERGED","created":"2020-06-19 11:08:00.000000000","updated":"2020-07-28 16:54:10.000000000","submitted":"2020-07-28 16:50:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736976-1595955028426-f4f80a02","meta_rev_id":"0bb1c2ae1cef126d9566c1650632a24a89e8f72c","_number":736976,"virtual_id_number":736976,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 16:50:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-07-28 14:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-07-28 13:51:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-07-09 09:26:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-08 17:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-07-28 14:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-07-28 14:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 17:27:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-28 13:51:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-28 14:36:32.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-07-28 16:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3835e6ebb36cbaf8aac0b30e68704eddf61419be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 11:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cdd6d99deb90703b5b84671c90ec661bbbc4848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:08:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e69340b6ee24cc4a3c2ee1a4f865f5e : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5c42583f414b7fbbbefc01447bd4b0 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426ec042d72346d5b11c824a99c1370b : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99472c09c6bd4d32875f62cbf88a0038 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1330b870f9b4c79b11c355c0cbf6348 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/711fb0c8d9d54a1e8a09123d9296d2a3 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33e1e045336f4342ad08f45df1f42e42 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7daddcbd4dc4ee897591e4a40599125 : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63c5f3640f0b4be3bde7741075fca708 : SUCCESS in 35m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4798668c82c449b2a2f17c4ea2ed580d : SUCCESS in 1h 15m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0e0110fcefce49258374f992b0358b74 : SUCCESS in 27m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/114326522fd74379a8158e0ee2d48008 : SUCCESS in 1h 55m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3891a82b021d440794d11d30ccee1f94 : SUCCESS in 19m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/67417c8e59814de5bbe86ae5e4ae8595 : SUCCESS in 18m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e05d0d907c38441898d11d8007a35337 : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6ce238518d5c636df0cf6f3475f0c372919d4381","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-08 17:26:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37a2bac29b1e0076c8b1bc13b6dbf3efd749cd91","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-08 17:27:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d8df1c8981f6a0ac7365e7402302fcc1e859b379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 19:27:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7d03d5616a74617b7f9f5695c01c222 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86853bf947494ce69cf9a6806dffa62d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c12d3fe0d6a24509bd9a812256abe2fa : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d32254dba0f48b6b07951e0a5874189 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f498da736844e8a5173fe55e896f12 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be9af3a4e02244bba2e6cdd06fd29973 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ddc6f6ea01947e9ac2d77061a7f015f : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db40bd905db14d58ac7865e2c831137f : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b446467d57c4d7eb3b00000af20a939 : SUCCESS in 28m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b96a75db989f4838af6a85a177d8a3f7 : FAILURE in 37m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f89c97c52a17417783f0d5b674978253 : SUCCESS in 22m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7492fb64ed814d378ef4ff767d40140b : FAILURE in 1h 49m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f53d1260f6014c38b010fc35d32323e6 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/05f5ed12a8b44e32ba8c98b48e9bf51f : SUCCESS in 25m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/63adfb86aced41028e99b53bf9795641 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":2},{"id":"65bfbff4e97336c7a4b2044cd58839e5021c5113","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:26:23.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"8638ebdb88938067b4bdb438bc8b20f06d11dc15","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-09 16:58:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7f368d784149ccb087a71a8fcf5768647c2f4897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 19:11:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f53077f04e47da9d1b1446b371f6a3 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d5a193a1974674a52b572453756d84 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb7a1642eaf44ae1b52247e56e9c4a81 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b9f76faec5c43ff8ec4dc8f3abad6bc : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e8062a95334ef18ef57763d7906e52 : SUCCESS in 14m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59d686afc15c45899014d2e0d18fdf71 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfec317620a3443b92283ac9dc329438 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62b8d747e3f0434cbd2bd8b5df4b70c5 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b2746b9a8824a7097c3ecde169b8dfc : SUCCESS in 38m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c042108bc3624e07af03fcedda1f9744 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/05960de6dc8e4d52bd1255549321542a : SUCCESS in 17m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c2704957e86d4914a71a04d70860d2ac : SUCCESS in 2h 06m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/29057c900d1240aba1555bbb21bbb0dc : SUCCESS in 22m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/35fe4a93fac14fe980119a3d4ce8d340 : SUCCESS in 29m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9724d3a7476941199c6b4ab934c1cb21 : SUCCESS in 12m 57s","accounts_in_message":[],"_revision_number":2},{"id":"91e2b6dcf06c7745018c9b4cb10dc1d3617b16f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-28 13:51:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d5ec74341063d47411c0d4fc087c9c8cd861b873","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-28 14:36:32.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e19bc5a88882ea024c3324fb467a0fb61c7687b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:37:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56605a5dfccf05bb03b0a739b46d0666489b381a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:50:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1a9ded1cd2abbf191d9c7ee50caf3c44466fc4bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:50:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a6d7fa78e6400ebf0f5e9fb3b042dc : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db05e065c344bb683db5532c6852d85 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/494c2ebda7a540f6a6b49810db5e58e4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c644fcb2c1140aeb9951d56aa53c014 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f65109857efc469297b5519a76b6e2a0 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e7bbb68111047b4a026152a4d3f71fa : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0ead3a5d47e4ebb96be5d2607183b02 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b9efa2529964cba8d8de4f2083fee21 : SUCCESS in 19m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6ae0ded0eadd48679ded4a48875bdc54 : SUCCESS in 45m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f5e460d3dfc5476699eb39f43a185bf9 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/adc86cf37181439b8a033df2baebca7e : SUCCESS in 1h 39m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1a6ff2dc5c114cb6bc8622fc397a315d : SUCCESS in 16m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/41c23de52ef24f31872b0a9d4fd10a71 : SUCCESS in 26m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/be9f5a2415074d49bd6fdb56e9e9b2b0 : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"97b9bd7d871530244e2274f0cb3b3f38a0d1fdbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:54:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22649eaac7c4b95a35c41633fcec140 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e21a3404d3bb4678ab445e157de3b42f : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b86fd9945cbee93eb5c74772cf6a66b87158db8","revisions":{"c1e00fff38e1609d7c2ce968233bee930db626ba":{"kind":"REWORK","_number":1,"created":"2020-06-19 11:08:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/736976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/736976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/736976/1"}}},"commit":{"parents":[{"commit":"da95fc897d9306f2fc082feff5497e9b6a206c05","subject":"Merge \"Fix kayobe baremetal compute commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da95fc897d9306f2fc082feff5497e9b6a206c05"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-19 10:00:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-19 11:06:28.000000000","tz":60},"subject":"Add support for custom options in static routes","message":"Add support for custom options in static routes\n\nCurrently it is possible to set the CIDR, gateway and route table for\nstatic routes. Other options cannot be set. This change adds support for\nsetting arbitrary options via an \u0027options\u0027 attribute in the route, which\nshould be a list of additional options to apply.\n\nDepends on\nhttps://github.com/michaelrigart/ansible-role-interfaces/pull/74.\n\nStory: 2007835\nTask: 40122\n\nChange-Id: Ib27deac4ddf4976a571b192ee5d1b8ca57701916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1e00fff38e1609d7c2ce968233bee930db626ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1e00fff38e1609d7c2ce968233bee930db626ba"}]},"branch":"refs/heads/master"},"2b86fd9945cbee93eb5c74772cf6a66b87158db8":{"kind":"REWORK","_number":2,"created":"2020-07-08 17:26:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/736976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/736976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/736976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/736976/2"}}},"commit":{"parents":[{"commit":"8d67d72fa45abb6e82eb810cbb9dec586669cb78","subject":"Merge \"Remove retired congress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d67d72fa45abb6e82eb810cbb9dec586669cb78"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-19 10:00:48.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-08 17:25:40.000000000","tz":120},"subject":"Add support for custom options in static routes","message":"Add support for custom options in static routes\n\nCurrently it is possible to set the CIDR, gateway and route table for\nstatic routes. Other options cannot be set. This change adds support for\nsetting arbitrary options via an \u0027options\u0027 attribute in the route, which\nshould be a list of additional options to apply.\n\nDepends on\nhttps://github.com/michaelrigart/ansible-role-interfaces/pull/74.\n\nStory: 2007835\nTask: 40122\n\nChange-Id: Ib27deac4ddf4976a571b192ee5d1b8ca57701916\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b86fd9945cbee93eb5c74772cf6a66b87158db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b86fd9945cbee93eb5c74772cf6a66b87158db8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
