)]}'
{"id":"openstack%2Fkayobe~737250","triplet_id":"openstack%2Fkayobe~master~I2e45876f89121d66cf03b50824bd8258517b88cb","project":"openstack/kayobe","branch":"master","topic":"inspection-timeout","hashtags":[],"change_id":"I2e45876f89121d66cf03b50824bd8258517b88cb","subject":"Allow to customise hardware inspection timeouts","status":"MERGED","created":"2020-06-22 10:23:57.000000000","updated":"2020-07-14 10:03:33.000000000","submitted":"2020-07-14 10:01:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737250-1594720870996-c502fe0b","meta_rev_id":"1c00b97a8eccb5f3d62f5f58b91eddcd6979ae78","_number":737250,"virtual_id_number":737250,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 10:01:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-14 08:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-07-10 16:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-14 08:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-14 08:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-10 12:40:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 16:38:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-14 08:03:24.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-14 10:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f177ee2f49b8321abe35143c72b52169cacf04a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-22 10:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"315e96155e935ff2f34cc333836c19093c5af3c7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-22 10:24:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a1bc08297effb19e0891c29b52827feaf228d356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:21:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50fae13b995449baa985d7f8e41dc690 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3d268003a2e45c8ac41719736b6cdee : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369111bdaae84d2e826bd9d383605e40 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c69d17c92e54caa853c693a946d7a9a : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e863a0d7462a40c5bc26e86139e328ca : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e86fc7179d741fca36841548960dfee : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1aa82894b8dc43bda6307cbed5d32abe : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f9eeeb8931f43da860982faaab79856 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f02fff96b29b41179d849b0b1aa8cd08 : SUCCESS in 35m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8077e97146cb4f3b8f9e61c1faeee06a : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5ab7b619835b4fc9b372ad23a39ac407 : SUCCESS in 19m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c8e39c90b4c4424c9905041bf80f3f18 : SUCCESS in 1h 51m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1cdf45240faa4e0093493038716be370 : SUCCESS in 19m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a19a1be45b56422eaaaf62d93241f6a0 : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e651b60dde3b4050a6977a6088ea399f : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0d736ce75b7c3f76c89b9370bdfe1587df34b3a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-10 12:40:34.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"e2f6c65610afd62bed313a28c26a6df5cedb975f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-10 16:38:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"552f31a321cf5fc058bff714e11a2272314e38e4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-14 08:03:24.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0657e931d66037e472bf4e8e194d5548e18f1950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 08:03:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"506b4e7a6d51c6e223502dbf834accbc43ac43a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:01:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb5a62b270cb4bd59199748eecc50e43 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af48089e723a4f1c9c77c81b9fa3df59 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d084678f90a46ac8ebb7bcadd43d028 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7b18d296bf47d989f95085ad35a8e2 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9107bda93824eb6afb671f32c339460 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5d3c9cdb29241c3aa44862244286934 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bef76acea4c14eedbc54b6224ce51320 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32d78089fafa4af5b66679972a6854dc : SUCCESS in 28m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4e7aa2e190a84bde97324deed894076d : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d19eee45642f4fcdadfbde3233dca5ac : SUCCESS in 17m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/94d955538bb24f9c8c43f667576fb347 : SUCCESS in 1h 54m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/683ee397c9aa4e64833e02a04027fbe5 : SUCCESS in 19m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e9f175f2330c43459bbf133f4ef3c88e : SUCCESS in 24m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/39a89284e4a249a78f090e97b4d2bf12 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6718369549548f10cfb44bbcbd4a5c5d08aebe23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:01:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"952eecccb0a5512dc5d8b7cbed0b12a5cef43614","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:03:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1030096eff3443b90b8aa51c3488754 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a47b52f63a402db71c457a3e873cc0 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db15a8012374d63fdae5b5b28729071a767442e8","revisions":{"8a5fa81833527c134bc1cc1a1d75a4159b20f30b":{"kind":"REWORK","_number":1,"created":"2020-06-22 10:23:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/50/737250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/737250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/737250/1"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-22 10:23:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-22 10:23:35.000000000","tz":120},"subject":"Allow to customise hardware inspection timeouts","message":"Allow to customise hardware inspection timeouts\n\nChange-Id: I2e45876f89121d66cf03b50824bd8258517b88cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a5fa81833527c134bc1cc1a1d75a4159b20f30b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a5fa81833527c134bc1cc1a1d75a4159b20f30b"}]},"branch":"refs/heads/master"},"db15a8012374d63fdae5b5b28729071a767442e8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-22 10:24:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/50/737250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/737250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/737250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/737250/2"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-22 10:24:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-22 10:24:00.000000000","tz":120},"subject":"Allow to customise hardware inspection timeouts","message":"Allow to customise hardware inspection timeouts\n\nChange-Id: I2e45876f89121d66cf03b50824bd8258517b88cb\nStory: 2007844\nTask: 40135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db15a8012374d63fdae5b5b28729071a767442e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db15a8012374d63fdae5b5b28729071a767442e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
