)]}'
{"id":"openstack%2Fkayobe~738065","triplet_id":"openstack%2Fkayobe~master~I1bb857bfca9e209bc6de30ae9852a4a08b2c7fb0","project":"openstack/kayobe","branch":"master","topic":"network-host-interfaces","hashtags":[],"change_id":"I1bb857bfca9e209bc6de30ae9852a4a08b2c7fb0","subject":"Fix network configuration of network hosts","status":"MERGED","created":"2020-06-25 16:52:29.000000000","updated":"2020-07-22 20:38:36.000000000","submitted":"2020-07-22 20:36:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738065-1595450182309-4aea1451","meta_rev_id":"de530ef94bbd49120aa6079368cd5568b69eac4e","_number":738065,"virtual_id_number":738065,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 20:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-22 11:05:53.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-14 08:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-29 08:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-14 08:22:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-07-08 15:37:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 15:37:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-14 08:22:24.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-22 20:36:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4decc76daa3f1428f6e8f0fecbb2f76e1c8ddbb7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-25 16:52:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72d0b4fe758d99f0aeb56897e8ab2462f8290dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 18:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/815651540c5945919c6261e83f7b1a0d : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491a17cbf2974df581e48ab74aff76db : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46127c5cc77c45e0ba73ddb5f41bc0b4 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1449d656bd674bf8ad87d6bd3a951780 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9bc58d1ad34dc1a5eeaee08c10f835 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24a2b130e5b443e98120e444c0607c0e : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/162f518f6ac44acfbba8e3605af7076f : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37cddb342d91451d99cf290affe2065d : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27c3af02c6b14fd099ab76f2e03732b9 : SUCCESS in 26m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3e6731a7a29a4186849dab6f37a9b80e : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f302f8d91b734c1ba9a7e7d2dfe41e9b : SUCCESS in 12m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b73515e4fea4f549abdda0431498468 : SUCCESS in 1h 55m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/80d1151ab1d14c77abd4958009361de0 : SUCCESS in 18m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e75ab2f961f84c9aa3a116f1163e8f96 : SUCCESS in 26m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a3ceb0d60ae440e5b91c50e012edb45f : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"585425f2c5463794052eca59586c556f46c06712","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-29 08:57:45.000000000","message":"Patch Set 1: Code-Review+2\n\nI think that referencing controller_network_host_network_interfaces in the advanced service placement doc was a mistake - it was really intended as a list of additional networks that controllers require if they are also network hosts, rather than a complete list for network hosts.\n\nHowever, given that we\u0027ve documented it it\u0027s probably the right thing to do to make it work.","accounts_in_message":[],"_revision_number":1},{"id":"4159361533d29130f2d3a7083d844b4ede13d351","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-08 15:37:37.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"cab7f0b8f9b1c2bd0fa10ecf6473aaa7fb1e44b3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-14 08:22:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"670c7a08f1196408120d4064a123fb8b6def9682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 08:22:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b23c0d81adebcfa8b04b75b954a2446ef7bd4c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:11:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c6d3e6d8bc4c67b0b7c3a8fcfea709 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba134b8673464e74a1f125d85324c8bb : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01abf127ea974160857b6f2adc43cb9e : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aea5d477c8a40ba89eaa14004f24e60 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f2bfc99458416a8cba228813c6f046 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8b546757fdc486fa620376a37a8ff1f : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be5bc0c38ebb421d8a1477ed5c286f80 : SUCCESS in 8m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0058216c82d14398a42875651c93a052 : SUCCESS in 40m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a3a3251ef59345f0b8f4c89f8e911e1a : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8dee1bb1ce1d47589b43fa5fbca1ac96 : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d938302a2ac14f09b4c14cef5a9cfdcb : FAILURE in 1h 45m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/710741be4a434dad92b211601a5bc9df : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b76caf427cba49da9f551850e33d22c5 : SUCCESS in 26m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/76c4bd6ef70945f483aee0a32d62950f : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d3a40ddaebc1d3e03c65b3193da128ab86466215","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-20 12:16:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60dbcc162f78a419f58b9bbf658d867bd3117380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 14:14:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb518eec91eb47a5a3520839643676db : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4f3532253343ffa75836b4be8d586c : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d7c8c5bb2b2408db557fda60a03ac60 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93f9bdf2c58941738386c356b6821c2e : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f1db9040e5484d89b0a756042d533e : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b9b85fd38f24e3fa35d543d4ebfa790 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbb46bbe4963435586b53eeca1fc8c1b : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b0ccc054fc548aeacbfa06f6b436ae8 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/023984d0a7a54a74bc1bb5717477d3ec : SUCCESS in 35m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c56fe8da54a74edcb81505d7db8ed872 : FAILURE in 26m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/13ae150bf9b341269e741ac75887e8dd : SUCCESS in 17m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/11027eda17ad4d23935b6e86a7902130 : SUCCESS in 1h 51m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/460540477b244c36b91a7bb21ea3d7c4 : SUCCESS in 17m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f7b962131fe0454b8c9c74b9b5dcaa63 : SUCCESS in 22m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0eb5d938bb194314b86719de935ac188 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dfd93c4e90c9ed76a573b26ba3d618b6d9adefc4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-22 11:05:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f25ff95492456a6f1d89ed18fa0c0421ac97ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e09f2c0b7af046e29cc81366b06d827e : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786b7f5b7d51446a8db061cc43f41210 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2bc674e19c94bf0b0e82a3a3398d39f : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1694faffe64e4d7bbf793f98b5dd0c42 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e1717e58cd4294b71cc27595612b13 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4affdfbb934032918777e591afa097 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21cb982420db404ab6c1a7e622e1fdae : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d239e5a83404468eb37aab3a5d21d463 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0bcf6e6f3d04a0089b2ee8ce14161fe : SUCCESS in 36m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/387746c98175469d8edbae2274d8a9d4 : SUCCESS in 1h 13m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3d26a9af99044dcdb02dcda56dc1e356 : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dc53f615be2b47c8bb2fda3cc52c1879 : SUCCESS in 2h 02m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1d85b27c296244fda4560f71da11f0de : SUCCESS in 19m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/37c15782c9ab4eff918018e8c4b0b38b : SUCCESS in 26m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8c87d4dac347485d8a1084ed2ca4b04b : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5b88c0ca6908d0d7e2344b722b77fd5d09833334","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:15:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"490efbfcb27178b5061285c19eb68e4d06ef323d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:34:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d610c35dde404a149c6fa8d1ed4c52f0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/609dd153a8c94acc89a8a905b093271c : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2448c5663ac4ab694776f9d8959bc6b : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ed322823e040ac8aa08f54843386c5 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a0c690a2acf4222a910323eb4092939 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce5954b9a73e412caf19a2216fd9262f : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5137149ced24c1fa2159d47004999f3 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/007c9bd6fa454eef96a0264b94ec0416 : SUCCESS in 36m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e01e4c929fd64d6199f8162a63de19d1 : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/84e325371557445d830712438576cfe7 : SUCCESS in 17m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/01d874f7b08448d5afac3d5a2ec33cc4 : FAILURE in 1h 03m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/16e51447360e4aed9d420c7aeb944831 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/519d4c905ba34900a124e8dd45afa84e : SUCCESS in 26m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/68ec97525e6a4f728c42ec5c8a72e88e : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1ce7e5b02ebef39614c3e95e60756e7bb9593db6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-22 15:12:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"935713d9c9b8ce83835551733f7cfe3b50179694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf49c25096a41d1804ce5fd71cdd5ef : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed57b5887d594f5c82c055385c874953 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a6be607c8a4693836e6f58ec61ece7 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8a72ee52a6047dabe9d18873e2fc8a5 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7828e4bd82340c1907a7133bf54b723 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfee8ee8451b4209a1cd833cc849fa4b : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/456e8c7157ac43e2856b1d486c556f44 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3264335f778a45b88371165db9af6a9b : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8277cf63b29a477a820fe7c262feb867 : SUCCESS in 30m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/09cc01e57455427aa834a360ee6250ec : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c191b19a65724bc4820c764a99e1517a : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8920254369514280beaefd37aa404394 : SUCCESS in 2h 04m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b27938895ba04468a5e25dd4a01da48e : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/198b338868a446da8f5c50fb4eb911aa : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3a1079fdbd85470997594c8539c08af4 : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fd6bd2f3dbdab67f5c94c77ffd9bdd4e3801b174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:40:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7ee3b10b079e04a02a305bfce919a41ee5cad8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:36:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3044172d2edf499598167bf4a491ba6e : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c82113e0400245dbb1e81a35c3b29f29 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c78290df3a94b1bb1db64a0c48a4a23 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e68c1c3d504cbf847598a57f7de9bc : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7193f9da6ea4b09a132aad0488652a4 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe0c952958014ab1a3bd41029ce43f33 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8a074efa0154d41892ed267d20fa5b5 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0adb4883295483c95f23fca88f7675d : SUCCESS in 31m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ca932238459b443c88ef223b1d6747de : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/020ffb95bc3b41508c81ba90b653c8da : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f2a96c24330b4c26ade128e131cb6989 : SUCCESS in 1h 51m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8901114722a54fd1a7abd9409b2398d8 : SUCCESS in 17m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b78c510f5ea24970a6054d814330adc1 : SUCCESS in 27m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7b363675e60346adbfb5d3abe25e0f56 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3aec8db2444ee4347bfdd73dd9f51859c4f50051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:36:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"60e23d18b9a07a69ff46da6b9ccf549937c1c463","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf701d6665448b3a5bb075382fdbc6b : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eb84fe44f8a4311a241cc3d26aa006c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58f26fb61bfda83a6b56095dd8570bfe31ac7cd3","revisions":{"58f26fb61bfda83a6b56095dd8570bfe31ac7cd3":{"kind":"REWORK","_number":1,"created":"2020-06-25 16:52:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/738065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/738065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/738065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/738065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/738065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/738065/1"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-25 16:50:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-25 16:50:49.000000000","tz":120},"subject":"Fix network configuration of network hosts","message":"Fix network configuration of network hosts\n\nThe Control Plane Service Placement documentation connects network hosts\nto networks listed in controller_network_host_network_interfaces.\nHowever this only contained public, tunnel, and external networks. For a\nfully functional network host, we also need:\n\n- the overcloud admin network, to manage the host\n- internal network, for services to interact with each other\n- storage network, for manila-share\n\nThis change updates the default network configuration for network hosts\nand adds a variable to define extra networks like for other hosts. It\nalso improves the documentation for adding network hosts.\n\nChange-Id: I1bb857bfca9e209bc6de30ae9852a4a08b2c7fb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58f26fb61bfda83a6b56095dd8570bfe31ac7cd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58f26fb61bfda83a6b56095dd8570bfe31ac7cd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
