)]}'
{"id":"openstack%2Fkayobe~738221","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I238aeefefa88dd584ce61ac9eb52d50552336ed0","project":"openstack/kayobe","branch":"stable/ussuri","topic":"placement-build-stable/ussuri","hashtags":[],"change_id":"I238aeefefa88dd584ce61ac9eb52d50552336ed0","subject":"Fix condition of placement image build","status":"MERGED","created":"2020-06-26 15:04:10.000000000","updated":"2020-07-06 17:11:50.000000000","submitted":"2020-07-06 17:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738221-1594055327595-64565078","meta_rev_id":"333ca0661115c70471bc1b623f8b96b846b64d3e","_number":738221,"virtual_id_number":738221,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 17:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2020-06-29 08:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 08:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 08:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"all":[{"value":1,"date":"2020-06-29 05:36:10.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 05:36:10.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2020-06-29 08:22:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-06 17:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37c9fead089e172770ee323a37987e735c1363f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-26 15:04:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f19fd48da4640a8934a75429279f2bbd50e5615f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 17:06:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6746b3d3535f49fa8363815cc0372d1a : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee82b37dac4d42599f6b57a43bdf9a36 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1ad6cb059594118991aa8a7efb0506b : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d48c80ab63be4550b24c8f6aa6d467fd : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e82cb113f724e71b311969290db98a5 : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a04bcb8df274cae99eda32d47d361a9 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e13db971bbc4aa795c4d3306be59ea8 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f960880426e741a19904615edc8d23f1 : SUCCESS in 33m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/87ad28a3286e4811ad2d600dce0d7394 : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fef8373fb50b4aed8f41fa4443324c93 : SUCCESS in 19m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2fc1c06fed684859a69e7a0f64e31aa5 : SUCCESS in 1h 55m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a6298985304f436084264686ce234135 : SUCCESS in 19m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3da2dbb2d2db414baa8163d088b1418a : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b8b10c938504539e5af98d8c5b45bc9e55f500c9","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2020-06-29 05:36:10.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"417699a4d5fb72f6fb73eab8ed5b79b55b950477","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-29 08:16:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8f84ecc537b262c975afd061b5a77edccd2c14f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-29 08:22:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1764914a4224fb7d2ff78f8d1ecb8f726a82ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:22:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98089633b0f2bcb0050dcd22af4a7529f7bdf2cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 09:38:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a549066a12a943278af019a6500b4a70 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad7454de409547a698f1f0feb4b9aa4e : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e96c4ec88944267971b45a75906ebcc : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047e1d319c7f4af3bae24f6c9f4e6114 : SUCCESS in 9m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0783166d21ee4451b927604d3096b1d0 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82152bf3126244f19cd3d62dc64f631b : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0775f7f691a04a1c88f130d97e70dd28 : SUCCESS in 36m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/88c98d96747948d99580beb390aad16b : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/192bf185852a46f48c663ba5ba4330d6 : SUCCESS in 21m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b56c312885840f09cfcb6fd36dada41 : FAILURE in 1h 11m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/936d159781fe47499e7b42340564a2d2 : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2302f9f01c944d89bfe6aa7e3423496 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"38522adb4f6988d0788989e4a8ac7c14c318061a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-29 09:43:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea3be98b1a5d013567b2de41ccf88ae2926745e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 10:17:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c8d07d576e14f67af0e227655efed30 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f97806189a043cdb932183e843da1d8 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24f8948bd8814865bfb2d182a8942566 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fff502aff8f4328a95d5382e2f7a520 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4755235f226424d8665b14911a57eb1 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7602af1dc5a04833afc77ce58371eb8f : FAILURE in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3205b4b9bb404cbcb3b6482e717b0010 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7789e4fb395940f2b9621ee6a971cc65 : SUCCESS in 33m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eea6caf5805d4dd78004c66282cb12b9 : FAILURE in 5m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3a7054084af74ec8b74811fd9f2dfca6 : SUCCESS in 18m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dcf456eaf51943a2911310e3243bb470 : FAILURE in 6m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0840410231bf4e52bb332f0f437967ab : FAILURE in 6m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7b148e0c77ed4315ba6f07d38821aa41 : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5388e9d4c9d77794b03cab4b29eda49866d7d0be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-30 08:32:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10c4c07172cadcb598ea512aa1ba5fc3f0665cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:46:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/761a8f16f8d545ac9266b552d0e395a9 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189fe058e4324abb861a35d8d36631ce : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6990480a0920414ca6dfed95b5838eea : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56672272fc344d38a418ed47b063ee8a : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c19d487b2c4301ad97d9c43d68d4af : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a655cb5223d34445898be956c378fa1c : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71dc73f5206c4426b43fe57a11612475 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18f4544c668d4bf39e3da9daaca782a3 : SUCCESS in 38m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/10ae1216e8b44f259e0d1f62d79b9ac1 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d774cd68e5454f2cafe0d9aa2ced3484 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6ecad17ab28a455d8be9ee65e436fc9e : SUCCESS in 2h 09m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/27e3ba669def42e4ac8d8d1609e10beb : SUCCESS in 19m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2cbb2a95944c475898fb8911c0fd5882 : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9a0a90150ecfe582f1fdf4b31a5f65d3a482fa16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:46:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef14330023e0def78840266952f625998fff3546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:44:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9760b92d301440b7ac2e7262461fa4f0 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9dfa9376183426a9fff2e99e346ee07 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67f3da19d8d74bca85f4ed77268d405c : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed51827708442c686ecfad68cfb1aee : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d90caff4104f43b2bd7535e018874afd : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f09d35c4a9d42458c1850c7b1e21be5 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30cf3c0f93184527a396b0dbee13289d : SUCCESS in 36m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f03314fd33da44fda82c9f3e39f23adc : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/88bd09228fbe422880c0a92ab1c1438c : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d044c9117984cb689dc610f1daf0add : SUCCESS in 1h 54m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eb627af217354323b5d13232d82fb97c : SUCCESS in 19m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e3e537cb9d804dffbb714ca820d54744 : FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"adadbc19a727c11469e4c51c1983e9d6a7b62bd2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 12:49:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4f06788bb3024e28abbfd91a61b4500224e7fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8a367365e8c41b6b76983589b6563b9 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a909395af52442397c66765424b0226 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6992ca8a7984a80a03760cf101ae523 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f2e90e3ff164e16b5e56f129c01dfbe : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e847e9eb994c8d971de33968acf965 : SUCCESS in 12m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2d6efb787804c959605b0f8629c1d75 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26f28064eed24776912af1f2a7c4f4cb : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82c7e6feac834c588813449950c5d537 : SUCCESS in 33m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4d3f867f64894750afc5c477dca936e8 : SUCCESS in 1h 26m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/64a67d9eb8b0408fa5dd9f58e8f7ef2e : SUCCESS in 18m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4bd79af0436444c1b8e1c128ea6971cf : SUCCESS in 1h 49m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ed6b53e579cd4b8f8db5448582bdd3c0 : SUCCESS in 21m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4934cd3abfac438ab89c2e333fe662f2 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9ad2b2964ea0df15e34f51f2bfe5da399f86ead7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dea836f9357402a9e45dad321aa238644f3b356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:41:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a652c4b31a4a4a1da40436f64c264269 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05513ed0a4704c7da3ef4230e4dfde36 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e42b318f0cc457da79ae4df2121648b : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859fb295c76e42bb8e8e40b68ce86675 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbc7b014edb24023b40cee590094c342 : SUCCESS in 9m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8455ea94942f46a19948259fb5cbaa8f : SUCCESS in 14m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0363475f76b4e528e343cf6bc9df4ec : SUCCESS in 36m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d22610d3522845d59438d3fd6d9b0757 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a319201a5b8a4d20b74e2010d72a439d : SUCCESS in 18m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/61e96d4e7d084938850f814b25c86290 : SUCCESS in 2h 00m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bbd15e2108dd488cb624e9bfb88f1ad8 : FAILURE in 19m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/60723e5a39674b4383a68c300195a7fb : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bba52e27f71901d0ebef7f7e7c574e816cac8791","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-06 12:31:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9fdb34be8d502628080a07aa4c1227d7455cd88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53415daa888a46dd9f088df83f50ce27 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f764f39c2547f59f63ec835cd538b3 : SUCCESS in 10m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74cc453aa074c98a9c0524960514ff4 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75375c818be34f92904b14affb82afb5 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43588e19d5dc44a2be1d4dabec219322 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f32742b6aba43f092463c583bad3f2a : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e00fa560b8c44537ae6110e89c18d684 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f726f1811c9043fc89309c53163406c0 : SUCCESS in 37m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3b39a125e7d940f0b1a71dd5c62a761b : SUCCESS in 1h 15m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3e9db34fd8ec46db8b042eff2ed4c384 : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7ff1bbdd4c634121aec5266da60527b4 : SUCCESS in 2h 15m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f1434ac3113a41df8b0eb213a9e8f04b : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b406ae12191e4ddaa1ea6ccd4a2419f9 : SUCCESS in 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"af04ca5e221a3dac0f5bd8ae8e3e292b57b80bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:49:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bfeb4616f4fe5cb18ea9eb12f08e1c885269ee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:08:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4efb0f41b2343e8a2e07a4e4096a497fc2305c3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:08:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ede29aece5402cbc88a16fe8ab5bbd : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f19cc14b77c4b63954b5985a2cbf781 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25b7812c35884637941f0f637663e245 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee5a7c2a28040eca138a6ed655741c2 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7ab230bc3174210bb0a4af88f6f5771 : SUCCESS in 7m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a0671d932084cc1a2f43672d547a9c3 : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd70950480d44a78ae0e34f44766ee8f : SUCCESS in 39m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7be0ae42e46145488a8e23faf8eb8aa0 : SUCCESS in 1h 20m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/105e465f5ba24f428b1137da6ec29b0a : SUCCESS in 17m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b391b291ee1849919af6277f1bcf6234 : SUCCESS in 2h 10m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/33ec8222056a4bcfa36b25e10e595681 : SUCCESS in 20m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/98bbea5b34f74f33affd98c9e9e6e5fa : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"384ce7e7041ae843daf4e15393a79e962ae9f925","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:11:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3045fa55ae144997aa95ba7bc21fe81c : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eef94567596cb30dcc866fd5a8cf162cdf9aa631","revisions":{"eef94567596cb30dcc866fd5a8cf162cdf9aa631":{"kind":"REWORK","_number":1,"created":"2020-06-26 15:04:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/738221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/738221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/738221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/738221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/738221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/738221/1"}}},"commit":{"parents":[{"commit":"9b75761a5f041b837cad42aafb4cb2cbeeb68045","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b75761a5f041b837cad42aafb4cb2cbeeb68045"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-24 08:45:41.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-26 15:04:10.000000000","tz":0},"subject":"Fix condition of placement image build","message":"Fix condition of placement image build\n\nThis should fix placement image build if nova is disabled but zun is\nenabled.\n\nChange-Id: I238aeefefa88dd584ce61ac9eb52d50552336ed0\n(cherry picked from commit 931fa4830d681281ff189bed15fbc9e0554428c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eef94567596cb30dcc866fd5a8cf162cdf9aa631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eef94567596cb30dcc866fd5a8cf162cdf9aa631"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
