)]}'
{"id":"openstack%2Fkayobe~738940","triplet_id":"openstack%2Fkayobe~master~If78edb12d699a47c911d28afbb365230a3021c7b","project":"openstack/kayobe","branch":"master","topic":"zookeeper-storm","hashtags":[],"change_id":"If78edb12d699a47c911d28afbb365230a3021c7b","subject":"Sync zookeeper feature flag with kolla-ansible","status":"MERGED","created":"2020-07-01 21:00:47.000000000","updated":"2020-07-13 09:34:20.000000000","submitted":"2020-07-10 21:53:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738940-1594418017503-f967fff8","meta_rev_id":"976f4961ce263421540b35334ba9e0210df7aba8","_number":738940,"virtual_id_number":738940,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 21:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-07-10 16:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-07-10 16:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-10 19:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 10:56:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-08 17:04:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-07-10 21:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"738f54b543f088a0867a0c25cc8aebc849b443c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-01 21:00:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16881e6045dd5ac1139073bdf486f8ae6821ee65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 21:51:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d249e280eb140e38632df863f190290 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1e2850241441e6a8c3194f63124147 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd6e6b1ac7147fc89777a439a918a8e : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee1aabbeec8f496aa62f5b436568eb2f : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bdffa026244f14a755bc8ab448fadd : SUCCESS in 10m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/092b90951d98481b8a25baa91e2b86ce : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/830bf623392240b3a83bd24bb865711c : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72432c21acff4147ac304a6dbe15e1ab : SUCCESS in 33m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/69d93872389b4fe5897698fac9077c2b : SUCCESS in 50m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3388ddbf75bc492b85f09e35a9038a62 : SUCCESS in 18m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/abb2196b0bc34869bb77047547a4cf50 : FAILURE in 9m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d1e085976461416a8b0a2135b1c2dc98 : SUCCESS in 19m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f163c1a477e444c2986c7aa7c2012327 : SUCCESS in 21m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/43d244e4396e4fa8ada325d29c4c8358 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ce84bb429f756aba7a4c6dac63301d1f52d23416","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-02 10:56:12.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3eb45fdaeb56845d54f42c9f807188b9f56fff6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:28:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31050b0cddb34a8cb3ba81004841be14 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d55ebe9172c4d2ba44a39553d5d1932 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07fb632ad0974bdd8628af18c5e0a55d : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca8dd408bed4906a513931cb837a4d0 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9715e8e3e92e4233b0acb996ba14911f : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d3bd7601cd841e2884f0fb1e2028a8c : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c052cc73c904a3c83cf6016685c22f5 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df94e066f0a74af8b9ba2fb59d6c471c : SUCCESS in 37m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/181ce7a7a32841d6b0fa13f44f06bc59 : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/983306dcd3c94261ba21fe230540377a : SUCCESS in 16m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/247943e9681445db866093ce32197cf1 : FAILURE in 1h 29m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bece2e07ca294fb487cf8f80fe237712 : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a57c598649654fd0b810c29e2d1c5eb3 : SUCCESS in 26m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fa46472a31d346b186770a20501f1004 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"52dc4253e3b97336ca305f2e4e0ab47a204bd218","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-02 17:09:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4785c82b0f56ea305bec4fd293f9d9a3e08ba1a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 19:46:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0b775b6121a4d149567182dc6da359f : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962e0dcafc5b432aa8f5e8e8fa53ee5a : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9523062320a24d6b84b92de2b0a5a92a : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81cdfd988aa34e38a499591b997b134b : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1038b460ecea46989428711e3f64dc44 : SUCCESS in 14m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b282348a103041cd9043b42d492ab062 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b790d25326a459597804f5e0df162c8 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81282d206ff8455c9770183459dcefa2 : SUCCESS in 38m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ccd5abb5d729470e800239ecd310b341 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7a2d4393ad02499f9db980616e189860 : SUCCESS in 18m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/61e721eb8dd74cf9bf35b3c495da197d : TIMED_OUT in 2h 33m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/053d860723704070bdde98b9a7d74810 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0ffa7549b5314455b13a580f297b2b1b : SUCCESS in 27m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f3ba0cf9d97045d3942bc3c93cfd35a1 : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"563a7fea9312d250e6bc0e80a6611f32d14231ba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-06 09:18:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59e45adbedcfcf7cd2db1098c7c42a3a58e15093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 11:18:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33b84255a1c242b9a4133056e3bc069a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88f0917383346599a646744274a224b : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/136fa70a763b44fdb009e71af1374f06 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca17852d01d4db4bf91e60531b6a820 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45c5a6658b7419180e930502b48e570 : SUCCESS in 8m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb119d32a658495088f41ded8e590c09 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1890ecacdec441aad003fce00446b69 : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5731ae224474f48b20e69fe70ef9ea5 : SUCCESS in 37m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1aa71a608d1945adac4e1cea53858890 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/df475bf4f3c14fb8b4ad994d457288d3 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/04a472c0d0664e1981d82cfd4d478679 : SUCCESS in 1h 55m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d788076d69574510a11e86602d1eacee : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3c7a2f7163e249dea36cd4b536e4115c : SUCCESS in 28m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/39c1ae60d9ed4d4ab2fceebf19319641 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a00767826ca764add6a4eb3dba45f6777cdd21c5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-08 17:04:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c4e4469b7572261d63b0bb2d347a0525a5445b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:06:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1a9c5341fecdc32f375cec863f267b11cbf9a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 19:07:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd5a54356eb438e8efe54fa4f46e5d6 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/235b85c4d6d1484594980af9409aa74d : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52e2c409f33d4b61bfb8625dc8bfa2c6 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6716684619574e9eb1490bcedeac1f03 : SUCCESS in 14m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38e1abb2272745ebbfbe86037c9eb855 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae3e953fb7174537aa9ec27bdacf7f9e : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/faa154e26dae461fa3dca78595469a6a : SUCCESS in 41m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/81d93395b80641629cc75b1f36ef3251 : FAILURE in 52m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cac18f703a7e4bfba36e677cfc92b305 : SUCCESS in 22m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0cb510764d6846de8c0898fba43ec080 : FAILURE in 1h 55m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b35dc164bd724348b9f56038cea2152c : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b25850548e744173b98284ff2ea9416d : SUCCESS in 24m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2e49fe9dd2a84abb9a7340ed16f5314d : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0072e8dad7c2593d9101a2a6964fbac7b0c6ce27","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:20:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"63befb8a7c827dd688cd3d2e0c12e3473fad36f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 11:18:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd6f0facd0c44e3a1747191f0b7390a : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fcfd9e6569a4b138eea380685f537c6 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2dde66732b14b5c8ebe39ea6369118d : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daa2b33a874649608e86da26b7e16f30 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52719cca18f64b339237a20c4ba6da63 : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d85c62020c14624a23db7d707d7f0a6 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9280e13bc2e34286934392b1f4885012 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db8917b889704fdcafa4d4f4feab90cf : SUCCESS in 37m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a44f8467bbf4489da285acce93dfa338 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0c99765dbef84b0d904862f6b904af83 : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ccd0f57ff7a8493e8fce21520b316ed3 : FAILURE in 1h 57m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/25cc2294bba44b21b77afc50b7a35773 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c6c27a67469f430b8c11376d3d4242a3 : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/89a16adebe644fd0be08ab3b58e887f4 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"75f82a6049ff410132f0a796b6671fc4a9e7b8b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-10 16:48:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"49a04dae78e431176ccb73639d8a9a9e781e21a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:51:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ff2b95dd9e46fd91b565cc83af2ebc : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7f8d5f75784a69ba8c0548c6d8fa84 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b61c1b536c499c854ce487d3a989f8 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3875d44d487644df99dd1a31c8835a30 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f020670a1754a79b1c57ba5e5c1ccba : SUCCESS in 7m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7fe433544641416489b478ae3dcd177e : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43f2afa19e774a73a2b218b47f95bda1 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a769b3585a0452698223bd3d0d4f0d3 : SUCCESS in 29m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/77e07197c709490a8f29b8af02205b69 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/20bb2daf164e485bafa550b57e7eb488 : SUCCESS in 17m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cb054d50634847bcba1a2c2c4a0eaf8b : SUCCESS in 1h 57m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/debb6c291e594f238709eedccfec3bbc : SUCCESS in 19m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/426ad776d622418092aaff412b2e1c1e : SUCCESS in 33m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5cb82c05d8034c64a0acb84bfba116d8 : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a646a333a2315185df919d3c56a8bc0d2c04c4dc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-10 19:56:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e39e55614dd343373645a1c734421c144ff9ef28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:56:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"825b843304fe018eaa8feb324822cd7863044c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 21:53:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a0a8b926142efe58212b93a1358087f715eb2ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 21:53:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b9bf70d4524c40a596220d81331d9b : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/790e69429a7641ecacfbe78e35432f22 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84bba6ff43be420f9894cc14e2f37cec : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7aa51944abc427693373e37cecc241c : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a481f38a6c484d3d80192bbe69e35a4f : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62fd726f47404976b196560d73c34e45 : SUCCESS in 8m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16a25a69a7a2449394e62d756f737a95 : SUCCESS in 30m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/941f27a4fc8a4563ba3c7e49ddfe7882 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4696a781d4684f0a8ea9696a5dc4711b : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/047eb192837544de8a18d352a4769532 : SUCCESS in 1h 55m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ee3bbd5db2fb41dd975ce10064c2fd2e : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/12dd92d5d6c44816a0694f33283f366e : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e43bb257fad742c9a5d2dee860e3b733 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"12ef15a38cd7a0c4088f2d5243d6d90f61cb9dff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 21:55:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c986742145f94d1097f62d2439cbcc06 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2},{"id":"cb1b7fd5f72ae1f7c97a750fc33ad2566732e5f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-13 09:34:20.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 84662a9ce738926879f03adfc56cfb50b1227312","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a4f3ed23e68d23120bc97a86372d02a4de204bb","revisions":{"8872729c00dce28d67650a1c468fbef604dbed84":{"kind":"REWORK","_number":1,"created":"2020-07-01 21:00:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/738940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/738940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/738940/1"}}},"commit":{"parents":[{"commit":"931fa4830d681281ff189bed15fbc9e0554428c0","subject":"Fix condition of placement image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931fa4830d681281ff189bed15fbc9e0554428c0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-01 21:00:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-01 21:00:10.000000000","tz":120},"subject":"Sync zookeeper feature flag with kolla-ansible","message":"Sync zookeeper feature flag with kolla-ansible\n\nChange-Id: If78edb12d699a47c911d28afbb365230a3021c7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8872729c00dce28d67650a1c468fbef604dbed84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8872729c00dce28d67650a1c468fbef604dbed84"}]},"branch":"refs/heads/master"},"1a4f3ed23e68d23120bc97a86372d02a4de204bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-10 16:48:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/738940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/738940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/738940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/738940/2"}}},"commit":{"parents":[{"commit":"335ed05a5d2d431f1c2b8754ec322c483f3ae7b5","subject":"Remove mongodb integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/335ed05a5d2d431f1c2b8754ec322c483f3ae7b5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-01 21:00:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-10 16:48:38.000000000","tz":0},"subject":"Sync zookeeper feature flag with kolla-ansible","message":"Sync zookeeper feature flag with kolla-ansible\n\nChange-Id: If78edb12d699a47c911d28afbb365230a3021c7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a4f3ed23e68d23120bc97a86372d02a4de204bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a4f3ed23e68d23120bc97a86372d02a4de204bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
