)]}'
{"id":"openstack%2Fkayobe~739802","triplet_id":"openstack%2Fkayobe~master~I01da8af2255e2ac4751279db712292f9b862adad","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I01da8af2255e2ac4751279db712292f9b862adad","subject":"Remove opendaylight support","status":"MERGED","created":"2020-07-07 15:42:14.000000000","updated":"2020-07-15 17:07:19.000000000","submitted":"2020-07-15 17:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739802-1594832695151-98bbf7d3","meta_rev_id":"abbd7345273f6fd64fb87242d730c0c9fc0b6b35","_number":739802,"virtual_id_number":739802,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-15 17:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-07-12 10:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-07-15 15:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-10 20:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-07-15 15:03:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 09:32:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-10 20:00:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-12 10:37:29.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-07-15 15:03:26.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-15 17:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a4db675ffae06986e10fed6593ab45b13f83a48","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-07-07 15:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b7e1578f697d9ea525cbea2e8c2ca49f91a4d0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-07-07 15:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dd4e441bceacdffcd5678b426a0a50892036f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:36:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d2efbfb139d4a65a467963f8aab45be : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5027d484df1f445f92d5ff4f1c2f7759 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d167d4ddb7b347c39d67da4cadc0db9a : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bab7cadc6aad4dc39c52960f8373d5c1 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df97eec83984d2ca12dc0d962c9e8f7 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d10e4468b1934f9f83b2a588c6888543 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/173e56410c104a0096d8941d75fccd0e : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de72571c72e5415c9a78b48a10dd5488 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/502e22da040243aea67f9c44015d2bbc : SUCCESS in 37m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b5582f50fcb44325aacf6231b0531af9 : FAILURE in 51m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2a219e520ebe49d59d178b9f5376290d : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/427b55af72414364845fea62bf20076f : FAILURE in 1h 45m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1f982ded5aaa40b0b311332528302c1c : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6600bbccc832489a98274a862d7f60cf : SUCCESS in 30m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a0a2ed5abbcd48aa9cb3a1da6a399b3c : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":2},{"id":"1ac4a8ecf6ff2ecc8aa336c4fc0c345cb029d54a","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-07-08 15:30:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2a4e543f608dd2255fa7d9711936d882a74a8731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:43:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b4246498aea4d7f8474439cfbb5252b : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbcac4f2e1a349a2ba27276842f4b810 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec56e5c3a85488aa6f1cf0b03a5bb35 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db14394caf2484c8e311d6515f31fc3 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c23f42be2724862843d44e2a4b18e73 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c9ecc59dbde4b698b5faec14dd4f223 : SUCCESS in 5m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1358e4b424e34ef895204ddd460a33c9 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/834bf2dcb5104263b1932b0a5a9326b1 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6a9214399ed41c6bf2fd42f59275f5c : SUCCESS in 40m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c028f8e74f3548a3a1bb8a029916f3b7 : FAILURE in 48m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2ef32524e96d4e1da7970fb90ae8dc5f : SUCCESS in 18m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7c0974da543e41d8a9f0a1b816d44407 : FAILURE in 1h 53m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4826dece000d4467a30e6359e8534f22 : SUCCESS in 20m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/94b26bfbd95947c2958356af5bc65f93 : SUCCESS in 28m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0b9f03d05da9460280446a79c7931aa6 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5cac9b48a490d0048558e792237dd44520a76720","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:32:32.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"a11868a1f207d2844d00f857e8e3b04c9bce6cfe","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-07-09 14:50:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d204573644d18c574b5e241ecfbe12ccdba44707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 16:53:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1eacab54489145b899939b978e4c139a : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81a78dfe85e484c8d7151cae23a3ed6 : SUCCESS in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b65fa8ebd0d49729ae36f0f1634fe7f : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18fefc3a030c49d086cc4050523e6887 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8485d5da38b415e9c0bd8a8a2f9ca68 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea266ee46c24e45bfea79fe65a210c5 : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f7d4f2649ae448d8b1439f000a7fbb06 : SUCCESS in 8m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f287d439ca541588154dda438813959 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3902426ea0c7426185b53e40fa0742be : SUCCESS in 39m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/18aea3e7796645e288d7e93ef4a0f7dd : SUCCESS in 1h 08m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cfbfb07c586442a792ef8d1a9fcff31d : SUCCESS in 18m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a700d3f6088a43cb90b34eae5650824e : SUCCESS in 2h 00m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f810f1c197414706a6a29427ede3c199 : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5749fb81fe8645b2a37e3910f584cb2d : SUCCESS in 27m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/070b6640353442188d65a83d29f0628c : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3456076103ac986682bd7b18849cd33f791f7572","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-09 16:58:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c56c4b67b29f29a0b4d77138bdf1c7fc44a9c10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 19:00:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfa5a9e042c849e8aee16d1ca340ac9a : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0616713716b44092a0f301cda92d9b03 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ad09163fd1b47f4ac5737b3a8ef1b10 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fc67bd6945c4453beddd68f5ef48f8d : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad19d14778d4cc89a482e992f4b2a25 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9df89ef07ef483d86b6f923d94059ed : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d46fb60d44642cc8b5e6dec31e62699 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31f2278e598248d28fa80ec4e9fb7ed4 : SUCCESS in 14m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15b138314ca3496cb5d26e24a655ce9c : SUCCESS in 43m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eea9880d1861475ca642c3cccecea478 : SUCCESS in 1h 24m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a6e0aad563194574943467f2143e1df9 : SUCCESS in 38m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1a40f1bca7114f8fbbe569a780a0aad5 : FAILURE in 1h 59m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7771b42d643640a6b56a338b9715559e : SUCCESS in 15m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bcefc538e5334cffaa0868e240328b3a : SUCCESS in 28m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a1e983b51b0a4451bd5711e3c5371607 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a604fa72b014952b4cbaa8e2b178d8f221732c2a","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-07-10 17:52:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e5f189efaf4532835be3e1565d7eb2f8b475b889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:54:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40294491de44486d8be65921eb8b0a7f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564cc50e864544d9b8524fad3c715164 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/103a34c4d44944e7bfc1b7cf118745ed : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/519e7ef1289b4532ac67b8da483411bb : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7434d0ba862c4664a9c18c84bf8546ad : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b769253c6c684103b32676c9c379d6da : SUCCESS in 5m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bd7b3d47001425b861031f07038780f : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4fb7df198474881af6127b421eb9833 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/511f9583abd04517bc4376376935d44e : SUCCESS in 39m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c8e19f4f7f4b4511ba61aff1b4e5bec1 : SUCCESS in 1h 15m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d663572f0d2a47c4a23ce4e363d4066b : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b53b9c47eb5a44d8acac3f17e53f6ba0 : SUCCESS in 1h 57m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9862fb39044d4087b8af0248ddf7a5b0 : SUCCESS in 21m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fad2f837b30540bdb083ec545a8a8bc2 : SUCCESS in 27m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/930185d223d5489180f13c6a1cc1b6e1 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":4},{"id":"4c15fd5a99579dd0b42cafd8962e4c0563c0eba7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-10 20:00:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c532effe232f6c2a327d143c777e718504ee2325","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-07-12 10:37:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e6248f28482c12958a24c0f1a21096f19e32f91e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-15 15:03:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c811af9afdb5163240f7d22a108487d730e78a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 15:03:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"346a9efc4e8213191171697d2331b3aa4bf9a7f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:04:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8712396a96aa4ab5b036df8af7fbfb47 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc55f4ee9260431fa83ef9d13625896b : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3798b368bf2b4f6d93ac20447e7bf71d : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e37b4376514b73ba1bfbf99fa8c3c2 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0894b6619f14ae4bb6c58c5d73a7926 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cee5d46ccad248559f1bf23a01428605 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b87e56f7866441e1b87bff5c620d03c6 : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/398783a410bf483684966b7916f29609 : SUCCESS in 30m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f9014ecaaa34140a7418cb8ed5f16db : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/878fa8205b4248568f0fee2a2be28b44 : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/801b64f96eac4b04b835b16983861b51 : SUCCESS in 1h 53m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c24dc7b013bb45c380d713514eb013b7 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a9b75b9fec4c40ac9aa7ea719f619d45 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7db7421a3d974596a57cba0b61395b9b : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"91d0f6937bbeb73759ed445769bc80dd59af7e54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:04:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"abbd7345273f6fd64fb87242d730c0c9fc0b6b35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:07:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c0e3da491e46e38353c0c1ffdd23e4 : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/631d86b7c05f4c7b9a2f07e08b85dd95 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3f6ffb9d83975abb4592db54531c26b4c1ad6383","revisions":{"d45621a765695c1a3b940142ca5c534982097e30":{"kind":"REWORK","_number":1,"created":"2020-07-07 15:42:14.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/02/739802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/739802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/739802/1"}}},"commit":{"parents":[{"commit":"8d67d72fa45abb6e82eb810cbb9dec586669cb78","subject":"Merge \"Remove retired congress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d67d72fa45abb6e82eb810cbb9dec586669cb78"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:41:12.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:41:12.000000000","tz":480},"subject":"Remove opendaylight supported","message":"Remove opendaylight supported\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I01da8af2255e2ac4751279db712292f9b862adad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45621a765695c1a3b940142ca5c534982097e30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45621a765695c1a3b940142ca5c534982097e30"}]},"branch":"refs/heads/master"},"071dea016c3b8ce786c5ffe0ad4d8acb446cc588":{"kind":"REWORK","_number":2,"created":"2020-07-07 15:43:03.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/02/739802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/739802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/739802/2"}}},"commit":{"parents":[{"commit":"8d67d72fa45abb6e82eb810cbb9dec586669cb78","subject":"Merge \"Remove retired congress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d67d72fa45abb6e82eb810cbb9dec586669cb78"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:41:12.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:42:24.000000000","tz":480},"subject":"Remove opendaylight supported","message":"Remove opendaylight supported\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I01da8af2255e2ac4751279db712292f9b862adad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071dea016c3b8ce786c5ffe0ad4d8acb446cc588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071dea016c3b8ce786c5ffe0ad4d8acb446cc588"}]},"branch":"refs/heads/master"},"d12c312a86a794dcf29fb0f0769c65d773b21d0a":{"kind":"REWORK","_number":3,"created":"2020-07-09 16:58:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/739802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/739802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/739802/3"}}},"commit":{"parents":[{"commit":"8d67d72fa45abb6e82eb810cbb9dec586669cb78","subject":"Merge \"Remove retired congress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d67d72fa45abb6e82eb810cbb9dec586669cb78"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:41:12.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-09 16:58:03.000000000","tz":0},"subject":"Remove opendaylight support","message":"Remove opendaylight support\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I01da8af2255e2ac4751279db712292f9b862adad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d12c312a86a794dcf29fb0f0769c65d773b21d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d12c312a86a794dcf29fb0f0769c65d773b21d0a"}]},"branch":"refs/heads/master"},"3f6ffb9d83975abb4592db54531c26b4c1ad6383":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-10 17:52:14.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/02/739802/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/739802/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/739802/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/739802/4"}}},"commit":{"parents":[{"commit":"335ed05a5d2d431f1c2b8754ec322c483f3ae7b5","subject":"Remove mongodb integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/335ed05a5d2d431f1c2b8754ec322c483f3ae7b5"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-07 15:41:12.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-07-10 17:52:14.000000000","tz":0},"subject":"Remove opendaylight support","message":"Remove opendaylight support\n\nIt has been deprecated in Ussuri cycle.\n\nChange-Id: I01da8af2255e2ac4751279db712292f9b862adad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f6ffb9d83975abb4592db54531c26b4c1ad6383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f6ffb9d83975abb4592db54531c26b4c1ad6383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
