)]}'
{"id":"openstack%2Fkayobe~740079","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I5b6d05e8e53b59bd88d0cfe41848ee3f14015521","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I5b6d05e8e53b59bd88d0cfe41848ee3f14015521","subject":"Add elasticsearch-curator to kolla-ansible inventory template","status":"MERGED","created":"2020-07-08 18:18:25.000000000","updated":"2020-07-13 15:34:27.000000000","submitted":"2020-07-13 15:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740079-1594654271052-167245bb","meta_rev_id":"3761cecef355d9ed90f8f4c3d7c007300cc19cfc","_number":740079,"virtual_id_number":740079,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 15:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-07-09 09:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-07-13 13:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 13:28:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-13 15:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0115ca2f42cd30c41dd4fe89b8bd3c76812b97ec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-08 18:18:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"052baa6b414a17b2b6437ba169f632bf38579c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 20:11:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d28176ce3e41f58469e6d4b3a0cedb : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c593dc667ae641e78f5cceceec666412 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/008b30d3195947b49ee78ff0e5607513 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94912fbcf6f477299759ede9b446c7d : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27ef1adcd82f4226ac08795c45d50dcd : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e275b053d56946fe8cdc2a9bd95100fb : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/429c76d38fc5445f8cfcd6d6b8cb4d85 : SUCCESS in 12m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e52ccc4f52da4b47852c515f3381934b : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b91068f99eb846ddbb9364c7192d83e1 : SUCCESS in 39m 05s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/3a606557dba54dda9be99d151cf0457b : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d927e965639c4083bb848c7fb3bdb2bd : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/774da16f3fd5496a9e66585936471902 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/361cde7bbaac49c8baaac755e255a3f9 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c973396ba4104bf1a14d1f0c1295af53 : SUCCESS in 1h 40m 01s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/31d6d34ad86b47e4a78477c5ac114fae : SUCCESS in 22m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cff8879a31624d8ab49e3531720609c1 : SUCCESS in 13m 58s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2198d6a2a12a4536b5f2fdaa1fd357cd : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"89a7e48954ef106b98e892fe2d318bc62a9355da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-09 09:19:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8db8fdd33ec6b74183812f7b7fd90fe1246af5d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-13 13:28:03.000000000","message":"Patch Set 1: Workflow+1\n\nApproving since ES deployment is broken without it.","accounts_in_message":[],"_revision_number":1},{"id":"8d50878475854bff701339a0f8f804e50b2e586c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 13:28:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36bc843ff2f83d2614dc0696b5f23bd5714bc530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 15:31:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d036a1bd0794a6eac0f6d68fae70b2a : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90cb69f49231414bbe71bc9495d4fb0e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb0a0d5f481401b9e026f31483605df : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51314af5123b4094b7e3c434a1a19d17 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b1b29752ce48efbf4667937acfd6af : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af500d20a36441daa885ad7d3fafcca5 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1e888bdade1401a8a46a69d5f9f45c1 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3b01daf7f294e6aa16002ba70aeff93 : SUCCESS in 32m 01s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/fe296c566907485ca7cc9efaee9af20c : SUCCESS in 1h 18m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c7efc5b6f5b743068729bf9e2a06a44a : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/8230a3311d874b0a847466c1fe416e21 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5a97c53b889a4e9d997dcc7fc9e893f6 : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4ee977423a7c457a9c6ccec16c0755c4 : SUCCESS in 1h 52m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a6383241280649feb697c1e24c428a4b : SUCCESS in 14m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/29498a704f5b4f459511c0c7183bb7fe : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9b4be33bb2c0485aa7cfb3decf140116 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0bec064e5c69afe7a03f21a66f794edd2c687acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 15:31:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3761cecef355d9ed90f8f4c3d7c007300cc19cfc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 15:34:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83564c3e8c2949a7a31b43425403f768 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"953879cb4a79aeae7f97263b4e2dadb3a6c9b4c3","revisions":{"953879cb4a79aeae7f97263b4e2dadb3a6c9b4c3":{"kind":"REWORK","_number":1,"created":"2020-07-08 18:18:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/740079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/740079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/740079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/740079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/740079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/740079/1"}}},"commit":{"parents":[{"commit":"3c32015baa54c6575d1d09e9bfc63679e00a2fb2","subject":"Use latest release of StackHPC iDRAC role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c32015baa54c6575d1d09e9bfc63679e00a2fb2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-08 18:17:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-08 18:17:53.000000000","tz":120},"subject":"Add elasticsearch-curator to kolla-ansible inventory template","message":"Add elasticsearch-curator to kolla-ansible inventory template\n\nSupport for elasticsearch-curator was backported to stable/train in\nkolla-ansible. We need to declare its group in the kolla-ansible\ninventory to allow its deployment.\n\nChange-Id: I5b6d05e8e53b59bd88d0cfe41848ee3f14015521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/953879cb4a79aeae7f97263b4e2dadb3a6c9b4c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/953879cb4a79aeae7f97263b4e2dadb3a6c9b4c3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
