)]}'
{"id":"openstack%2Fkayobe~741241","triplet_id":"openstack%2Fkayobe~master~Ia7655f28eb819a22aefb9560b8870556f56dd558","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ia7655f28eb819a22aefb9560b8870556f56dd558","subject":"[DNM] test CI","status":"ABANDONED","created":"2020-07-15 15:22:33.000000000","updated":"2020-07-15 19:50:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9188b1b99b4ad1196b3591f437c4a9354a010ad6","_number":741241,"virtual_id_number":741241,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-15 17:15:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 17:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf30ec28137a2f7c400d4464bb0d06f16a66402b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce0d04df3410d5fc1cfb498c932b6fda04cfadf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:04.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7bcde72fdafcb1233fc92415da06426454bcf220","accounts_in_message":[],"_revision_number":1},{"id":"3372f38252e7b6efefe6810ceef4341bed9ecc70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:15:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b75737d88b748748495c7f05fe3c8f7 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59fbe47971aa4cee887048c41dd9270a : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c07e73a7fc94c18b0516d4ad2726985 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c429035f2c8549d99cb1ce88d9b53914 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfbf3f2152a46fa844461cf311f79e9 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f670f839d3c24eee83d77dad3a44db26 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/297c141ca4c04d2cac888db1b7d05280 : SUCCESS in 17m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d3739eee59b46b29e829582e8358cb4 : SUCCESS in 40m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c9eaaa01c4fc480f843fd8f1fa62ba1e : SUCCESS in 52m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/060914c3ff724b36a0b018816cddd004 : SUCCESS in 17m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9f0e249fb94844568152f501df0a1545 : SUCCESS in 1h 50m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f41a0596e04f4ae1a9f54a9cbb006dcf : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/969240a059134ec1b0800a6e46ff7576 : SUCCESS in 28m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ebefb66bd32142b2b3129645047734f6 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9188b1b99b4ad1196b3591f437c4a9354a010ad6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 19:50:12.000000000","message":"Abandoned\n\ntests done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e984a7bd18d59f11c9feab3231ed68e8b73d91f8","revisions":{"e984a7bd18d59f11c9feab3231ed68e8b73d91f8":{"kind":"REWORK","_number":1,"created":"2020-07-15 15:22:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/41/741241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/741241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/741241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/741241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/741241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/741241/1"}}},"commit":{"parents":[{"commit":"fed87a1e9d1ead2daa3dd0e6e018ae4cedfa821f","subject":"Merge \"Allow to customise hardware inspection timeouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fed87a1e9d1ead2daa3dd0e6e018ae4cedfa821f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:22:19.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:22:19.000000000","tz":120},"subject":"[DNM] test CI","message":"[DNM] test CI\n\nChange-Id: Ia7655f28eb819a22aefb9560b8870556f56dd558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e984a7bd18d59f11c9feab3231ed68e8b73d91f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e984a7bd18d59f11c9feab3231ed68e8b73d91f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
