)]}'
{"id":"openstack%2Fkayobe~741244","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ia7655f28eb819a22aefb9560b8870556f56dd558","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ia7655f28eb819a22aefb9560b8870556f56dd558","subject":"[DNM] test CI","status":"ABANDONED","created":"2020-07-15 15:23:18.000000000","updated":"2020-07-15 19:49:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ecaf83c4f80f2c5c04ff3bffeb3820abc1db58d","_number":741244,"virtual_id_number":741244,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-15 17:36:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 17:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6736b5bae950420c1d3d5a66a0ee0e925778244","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"39040e3daee619f2a543f7455da0457bf028d2c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 7886e1b83d777534dd23bb1b4f6514809779bdc2","accounts_in_message":[],"_revision_number":1},{"id":"7e9222afa450c63b547ad838b52c03d7cbd6436a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7409ff409bec461082d0091bc081dcf9 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c9d718faa6d443bcb9e4ac220f0fb826 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a05f84263c6c46eaaab3a8d76294c7d9 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23546a38ade6429f92899c2f00e4729e : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa55ddff1f4442cf8db389aad392df56 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2847968d1a35462481816e994832fec1 : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab495eee728742e5a9602687f90c31fe : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/133240d174764c049da3173408b652d0 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df7fb90b22094379bab4db9c57f10033 : SUCCESS in 39m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d4b8379051894359aff63f8f0a568356 : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a55187b8c8fe4a53b368dc0e3b917204 : SUCCESS in 2h 04m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/034854db51514eccb6262a626cab0c0f : SUCCESS in 22m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f83a178b15374bf4914268e7381af19f : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6ecaf83c4f80f2c5c04ff3bffeb3820abc1db58d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 19:49:57.000000000","message":"Abandoned\n\ntests done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a49bf9756502181ca24316205584363d19b1d450","revisions":{"a49bf9756502181ca24316205584363d19b1d450":{"kind":"REWORK","_number":1,"created":"2020-07-15 15:23:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/741244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/741244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/741244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/741244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/741244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/741244/1"}}},"commit":{"parents":[{"commit":"399e9eacf743c7b7a38a10644418d51c226dfebc","subject":"Merge \"Don\u0027t disable SELinux if it is not installed\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/399e9eacf743c7b7a38a10644418d51c226dfebc"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:22:19.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:23:18.000000000","tz":0},"subject":"[DNM] test CI","message":"[DNM] test CI\n\nChange-Id: Ia7655f28eb819a22aefb9560b8870556f56dd558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a49bf9756502181ca24316205584363d19b1d450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a49bf9756502181ca24316205584363d19b1d450"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
