)]}'
{"id":"openstack%2Fkayobe~741245","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ia7655f28eb819a22aefb9560b8870556f56dd558","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Ia7655f28eb819a22aefb9560b8870556f56dd558","subject":"[DNM] test CI","status":"ABANDONED","created":"2020-07-15 15:23:22.000000000","updated":"2020-07-15 19:49:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00a4f4792f16a5b16683582edc461595c017d700","_number":741245,"virtual_id_number":741245,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-15 16:41:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 16:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3f931bfef2c10450bd20146a5a180c247831ee9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"77c3c7ca59167b29b225b7e5a573def8702b0e07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:27.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 479c2980ce8352ecd5a9d205d91216718afeaf3d","accounts_in_message":[],"_revision_number":1},{"id":"31f62e9b249e54e0fe3e0f1205d82bf69a854b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:41:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2656807c03bd44019e27d7d21527b5ae : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/77606dd2433e4d539b094b5fc1cd7b0f : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4f03be4040489d9e21f0d28f25f13b : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce28a661cde1465ca11d4f69ffb07452 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6844ef2d6d44368010ec1f587dba57 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5712a133aa0e452fbab111b87617b575 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76060fd1ab514b1780c1f235f89117ce : FAILURE in 8m 54s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/109b49fe3fa04d87a04aeceea579529c : SUCCESS in 33m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d3b64f145c34415595cf0e6b2812f9e1 : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e0ddfa78790541108041a628f7730f18 : FAILURE in 1h 04m 58s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/fca940d823ec44fd82765a60f0dea58e : SUCCESS in 20m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4bd1f5089c6345e8bb497125f46f4c01 : FAILURE in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00a4f4792f16a5b16683582edc461595c017d700","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 19:49:46.000000000","message":"Abandoned\n\ntests done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7886e1b83d777534dd23bb1b4f6514809779bdc2","revisions":{"7886e1b83d777534dd23bb1b4f6514809779bdc2":{"kind":"REWORK","_number":1,"created":"2020-07-15 15:23:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/741245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/741245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/741245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/741245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/741245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/741245/1"}}},"commit":{"parents":[{"commit":"36b8438297afee5d08945555dfe5c74a90fd35ee","subject":"Merge \"Don\u0027t disable SELinux if it is not installed\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36b8438297afee5d08945555dfe5c74a90fd35ee"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:22:19.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:23:22.000000000","tz":0},"subject":"[DNM] test CI","message":"[DNM] test CI\n\nChange-Id: Ia7655f28eb819a22aefb9560b8870556f56dd558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7886e1b83d777534dd23bb1b4f6514809779bdc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7886e1b83d777534dd23bb1b4f6514809779bdc2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
