)]}'
{"id":"openstack%2Fkayobe~741246","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~Ia7655f28eb819a22aefb9560b8870556f56dd558","project":"openstack/kayobe","branch":"stable/queens","hashtags":[],"change_id":"Ia7655f28eb819a22aefb9560b8870556f56dd558","subject":"[DNM] test CI","status":"ABANDONED","created":"2020-07-15 15:23:27.000000000","updated":"2020-07-15 19:49:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"332df0b6537a752d059309a87f5422365311e3e5","_number":741246,"virtual_id_number":741246,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-15 16:01:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 16:01:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0d9fa70fadf0bcc2895fe8d8ca4a627299db1e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 15:23:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"9d62a062cbced61472e716d6d8d8ea644125e98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:01:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/61b60413ca744e559307bfcb68bc74bd : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86a7ae8dbee484eb3b3aeaa35c2154c : SUCCESS in 8m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/885159bcdd054f77b1def45e268de0c8 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be6d8caef8e4a6dbb71dfcdb33182e0 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f15b514ddf124302aa87468b8a6cdd46 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88f735416c3d4766ab40e15f80f70bfb : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21545171c86b45c88f8c0f4892357da0 : FAILURE in 7m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f785f0fdc1d4eb5a8ba6c89d8a44cc1 : RETRY_LIMIT in 3m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/26e72f1bb38f4b86a63fe326149ecf58 : SUCCESS in 32m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/71b41fd3175748e99804da8e576f2be3 : FAILURE in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"332df0b6537a752d059309a87f5422365311e3e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-15 19:49:41.000000000","message":"Abandoned\n\ntests done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"479c2980ce8352ecd5a9d205d91216718afeaf3d","revisions":{"479c2980ce8352ecd5a9d205d91216718afeaf3d":{"kind":"REWORK","_number":1,"created":"2020-07-15 15:23:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/741246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/741246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/741246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/741246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/741246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/741246/1"}}},"commit":{"parents":[{"commit":"c5e04c838925f8a04cea68edd05bdf85aac189d3","subject":"CI: fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e04c838925f8a04cea68edd05bdf85aac189d3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:22:19.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-07-15 15:23:27.000000000","tz":0},"subject":"[DNM] test CI","message":"[DNM] test CI\n\nChange-Id: Ia7655f28eb819a22aefb9560b8870556f56dd558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/479c2980ce8352ecd5a9d205d91216718afeaf3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/479c2980ce8352ecd5a9d205d91216718afeaf3d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
