)]}'
{"id":"openstack%2Fkayobe~741950","triplet_id":"openstack%2Fkayobe~master~Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5","subject":"Sync kolla-ansible inventory for Ussuri","status":"MERGED","created":"2020-07-20 13:57:05.000000000","updated":"2020-07-30 09:23:22.000000000","submitted":"2020-07-28 22:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741950-1595976295106-b3c41433","meta_rev_id":"b4ce6e250f0f3b8003c070265116149d5dea118d","_number":741950,"virtual_id_number":741950,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 22:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-07-28 14:37:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 13:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-07-28 14:37:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 13:51:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-28 14:37:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-07-28 22:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03b1831453c7975843d7c60aeb2e22488bec5d82","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-20 13:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3b8bdc33ba20103003504e9fc5556f53dd8233e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:17:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10127dd6b95a41f3a640131b1f53af19 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844b795b916d444983a0eae679bdb18a : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f9c57fdbe2445f904ba1bef7263574 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60df057af59141e8bb5897252fc0d913 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d1ff22b7304770ad631599870692b9 : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7094576c22134c31a291661368f3196f : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18754b8d9a2c4c5f879e4dcd9c49acc2 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db2f93737fc349a7940506c65d261dee : SUCCESS in 30m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bd0be5c681a743cf8bcc171cfd0af4e8 : FAILURE in 46m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0f960e8eeef9415c9b2f36a6b7a6f273 : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d1914b31e2a6409a965f09713458cfee : SUCCESS in 1h 17m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2e0aa73c51fd409ba5a8ab2c3e995bcb : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6273489494f8467f8bbd157453ba974c : SUCCESS in 26m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4f90afb3cbb2418882bef5e56714815c : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c43a659c07b50f3ba70aae271921938099de2c62","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-21 07:43:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9babf128f8649a435fdc9d26cd5a165ddf3d9ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 09:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a69037ea88734c229b0900006a12d13d : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860ca42dd3874b7e867bb04ab8488ade : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e92a73274249413db9732d74524e2dbb : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d0f799dfc12412b870ed5d1e3872bdb : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f34dd2809b4f738dfb93853ba5e016 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a26bf15655c343c9ac1968f0d99aa58a : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b2ab4a7914a491ea141d702729d8bb6 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/325fa9448d8748da926773f84df0737c : SUCCESS in 37m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3b47bdb7e719464bb09d77b1bf82e319 : SUCCESS in 1h 18m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6522e3f6d89349d4a7f26ace5a40f8b8 : SUCCESS in 19m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6271ee0094a249dc995ccd57c4d3c626 : SUCCESS in 1h 56m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9e6d308f3274402fb38cb21415fb6855 : SUCCESS in 17m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a84c1bbd10b34e19a4cceb98bd15cdd3 : SUCCESS in 28m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/89c3011244ed4ea3b1129c18bc232399 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3ef38905ac0b9d3e87c1708f2a1000b8c84a5238","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-28 13:51:52.000000000","message":"Patch Set 1: Code-Review+2\n\nNote for reviewing similar patches in the future:\n\ncd kayobe\ncat ansible/roles/kolla-ansible/templates/overcloud-{components,services}.j2 \u003e /tmp/kayobe-multinode\ndiff -u /tmp/kayobe-multinode ../kolla-ansible/ansible/inventory/multinode","accounts_in_message":[],"_revision_number":1},{"id":"69b4c7eb56b27e63bb47341f4175cf51b88663f2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-28 14:37:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2929e4306e5f5b1c862af603b3c36f23107e5417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b7a1379c36827f1aa8147cb470f045779bc432c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:50:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e1bb20f777402f8d64dc80a87027af : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f147d829e2d743809a7340b2b6c117cb : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0326265a544143feaae01af362d6217d : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f79a0c484364351a250d911a1e87800 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7190d9e7abc437dae7f2d0d46e7ed48 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a2e76cc72fe473780489b2117a982ff : SUCCESS in 8m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c145717884e247e4a559a74050b0ac51 : SUCCESS in 22m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a84abdddeea9464687b96af1fdb7df4c : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eff6e27f77a7438f8d2efd06b7d90d9e : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/90a3dafb95e243a493c4befd48645273 : SUCCESS in 1h 25m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/50488cadd6894abebf20a60066a01cdb : SUCCESS in 17m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/03c78a631e604ce29ea2cdf5537aaf35 : SUCCESS in 29m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f348cb4c280040cdb68a58f8a9516bd7 : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"00b885400ce3d1e91f1535b7861e3bef555c83eb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-28 18:09:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"641165425c331c25b6c588cbce4623131e1572ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 20:33:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369399ee21d041ccba144dd8802d96d9 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d61d4748e64afc82d6118971adee9c : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1dbc706370149b5abe06b5b48b76927 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8451bffcf2149ed9458f54986a54e8a : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849e0e598ff04c9bb7c3ab09b3f40b87 : SUCCESS in 5m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23dab81186bf494882edde342875c615 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d955918a7d0841c496dc857cde3db1f2 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1562c7703b074326908e57f0d448d39f : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5f029e0034c43c2a492a1b690a64493 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d3c85e0a0eae4b09bf8b34676653a669 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/461b30441ec24790aa81447224cf89a7 : SUCCESS in 2h 04m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f6844ca9a2a4a3d8c62e1a50f9c7d19 : SUCCESS in 19m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bb02495d82b04963ba11b710a41dcc94 : SUCCESS in 26m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c9f8fd67b4e04beba0d06cbffa67f3b2 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fef58f07bd9f114d978e6bb9d2e70dbf2a0617ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 20:33:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"353d458c9024f7ebb1a1cb82c5d0873ada0a83e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476fad692e9e43a089bc7ba267a959b6 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fa9a8d742234dd9aacd06298b7508b1 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7da65fe1da545e4b0386df7f0f88c16 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb40065c11645f5a9d9930bbdfdb910 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d64c6583183c474bab0f0ced99ef6a66 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b65d6ab00fff4ff89b57c9297f547335 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb11c97988ff498aa3a120c8a0db78d0 : SUCCESS in 18m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6e8c6a9bf890445f8df758f7c905e96a : SUCCESS in 46m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e061b6782f184aa5a3b8ff941e43f0cc : SUCCESS in 17m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ab9aba4d60b048a0b46a3da0fc755410 : SUCCESS in 2h 04m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c154c89dd3f4413da16becff838d84ff : SUCCESS in 19m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5761f2ff72ab4fb5843c70167f95479d : SUCCESS in 31m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4c01d3c142984a9382afc164fa3dad0d : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5ee9da789ca7d3a161e03dbaf1b6ab12e63e46b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:44:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f07cfbf7c5841b946cacbb868aee9510d100f88f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:46:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a49d5116c34f7db8b885ece200b867 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b4ce6e250f0f3b8003c070265116149d5dea118d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-30 09:23:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6c65cbbcdde92e03726f443b09ee187ece851859","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"936f393d6329982efbfbe876237a96f1a40c92b0","revisions":{"936f393d6329982efbfbe876237a96f1a40c92b0":{"kind":"REWORK","_number":1,"created":"2020-07-20 13:57:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/741950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/741950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/741950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/741950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/741950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/741950/1"}}},"commit":{"parents":[{"commit":"2052b4b942cf39b7882fd7821f5f7dd39171fa05","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2052b4b942cf39b7882fd7821f5f7dd39171fa05"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-20 13:56:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-20 13:56:59.000000000","tz":60},"subject":"Sync kolla-ansible inventory for Ussuri","message":"Sync kolla-ansible inventory for Ussuri\n\nBackport: Ussuri\n\nA few services were added at the end of the release which we missed.\n\nChange-Id: Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/936f393d6329982efbfbe876237a96f1a40c92b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/936f393d6329982efbfbe876237a96f1a40c92b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
