)]}'
{"id":"openstack%2Fkayobe~742453","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Icf64b15e58c84f31a11cdefcdb79558b58737da1","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Icf64b15e58c84f31a11cdefcdb79558b58737da1","subject":"Update seed VM image to CentOS 8.2","status":"MERGED","created":"2020-07-22 14:45:48.000000000","updated":"2020-07-23 17:30:25.000000000","submitted":"2020-07-23 17:28:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742453-1595525283628-dfe663bb","meta_rev_id":"fd2e08958117e7f6a8fb2a4c7a392485fde65731","_number":742453,"virtual_id_number":742453,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 17:28:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-07-23 09:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-23 15:25:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 15:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-07-23 09:29:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 15:25:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-23 17:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06c93d39db59db1a1926c1e6ad098d4245919b9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-22 14:45:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e1b32de1e90fbb7c8641e0083d79a685452edee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:35:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd88bf304a694ded8164676d3c5820e6 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e0aac2ba464b679dab548669f16370 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49545f366c744577a047f1255f7a480e : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bbc6599dbf84d9aa81e92bd16de7190 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e340794ed78d467a82f2718d7e266dcd : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f402a1ee130b43b1b3e5147a8ce215ad : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3377b3be626547b2a43b4b7ca18ba4b6 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/caf3279189cb4fff9c5243d5fdc99723 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2953d9a8bb8a43089a043bdb35c2ef12 : SUCCESS in 28m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c14231066b594fddb15c6c1ebf8ed6d9 : SUCCESS in 1h 19m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f3547a7d40774e83bdefc226c12c207c : SUCCESS in 18m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af70f2e45a184ae99e103dd7f84a900a : SUCCESS in 1h 43m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f69b646ca5924e68b9554c9f8d17279a : SUCCESS in 21m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/19787b93f1424cdabb1ba43fc7635d3e : SUCCESS in 27m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e1635b6d4c59471bb9560832e3efd281 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"366e71c177f91b974204695d362a680cd0afa004","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 09:29:15.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"5bc1d1a0e9743aa08c2429a3641a48723079992a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-23 15:25:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f17dc0e7602e0a52b498a9c721ea6c67accee01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 15:25:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d19bc410f166990a7f72f8ec5bef9c8b3877754d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:28:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a645c0ff072e9f8aafed21b03bd3cc9c22a2998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:28:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d58cc521a71474a8c1d61caa2c5015c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12871bf45d834b65a169c8a907f0e196 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05379db014804de18972092550f11b60 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda08ff3beb9406db100f2f6207e7d7a : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c59dae91e84213b6c1188e3caa32e9 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a4eff0a981f4538bfcd0c2a3d1c0a3a : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/edb69f5658084b228ec59692df5b2453 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac72a27166d14dcf9c21aba133255b7b : SUCCESS in 31m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2fd181df92f844b39ad0298831a19fde : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dfbd62f6e4f341ddaec59e958346673d : SUCCESS in 17m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6ddb2ce9cca54ff6a2c3d39f0f06e629 : SUCCESS in 1h 55m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1945a8e01fde4676a7e15df0aaa2fc05 : SUCCESS in 19m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4a0c470d1d2b47d6a358941849536366 : SUCCESS in 29m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2350ce631a384a92954e05ee864aaace : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fd2e08958117e7f6a8fb2a4c7a392485fde65731","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:30:25.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c94933d599c4fac91e21a71aefebcde : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23cebd30c27748dcb19945019ddf0d60 : FAILURE in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bbea8ebbbcc763406ccef195ab546edcada8502","revisions":{"9bbea8ebbbcc763406ccef195ab546edcada8502":{"kind":"REWORK","_number":1,"created":"2020-07-22 14:45:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/742453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/742453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/742453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/742453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/742453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/742453/1"}}},"commit":{"parents":[{"commit":"84662a9ce738926879f03adfc56cfb50b1227312","subject":"Sync zookeeper feature flag with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84662a9ce738926879f03adfc56cfb50b1227312"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-21 13:13:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-22 14:45:48.000000000","tz":0},"subject":"Update seed VM image to CentOS 8.2","message":"Update seed VM image to CentOS 8.2\n\nIf using a CentOS 8.2 bifrost_deploy image on a CentOS 8.1 seed host, we\nmay see an error such as the following:\n\n    TASK [Ensure the overcloud host image has bogus name server entries removed]\n    fatal: [seed]: FAILED!\n\nRunning virt-customize with more debug in the Bifrost container, we see\nthis output:\n\nqemu-kvm: error: failed to set MSR 0x48e to 0xfff9fffe04006172\nqemu-kvm: /builddir/build/BUILD/qemu-4.2.0/target/i386/kvm.c:2695: kvm_buf_set_msrs: Assertion `ret \u003d\u003d cpu-\u003ekvm_msr_buf-\u003enmsrs\u0027 failed.\n\nThis appears to be an incompatibility between qemu/libvirt in CentOS 8.1\nand 8.2.\n\nThis change fixes the issue for new deployments by updating the seed VM\nimage to CentOS 8.2.  Existing deployments will need to update packages\non the seed host to match the bifrost_deploy container.\n\nChange-Id: Icf64b15e58c84f31a11cdefcdb79558b58737da1\nStory: 2007942\nTask: 40407\n(cherry picked from commit f618a2c4b823023c45aeb75df498b514f9483782)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bbea8ebbbcc763406ccef195ab546edcada8502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bbea8ebbbcc763406ccef195ab546edcada8502"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
