)]}'
{"id":"openstack%2Fkayobe~742494","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","project":"openstack/kayobe","branch":"stable/rocky","topic":"monasca-defaults-stable/rocky","hashtags":[],"change_id":"I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","subject":"Define default variables for Monasca config","status":"MERGED","created":"2020-07-22 17:12:59.000000000","updated":"2020-07-29 17:15:47.000000000","submitted":"2020-07-29 17:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742494-1596042853572-76a6aaaf","meta_rev_id":"efaea2c8269fcb25c4a4ab1c8432df74644188d0","_number":742494,"virtual_id_number":742494,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 17:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-29 15:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-07-23 09:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-29 15:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-07-22 17:12:59.000000000","_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 17:12:59.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-07-23 09:23:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-29 15:10:27.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-29 17:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acac60efb2334e7407b24bda82d413dfd961fc90","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:12:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"18d4c8502e1e27136bea90b87b146c9fcb40fa6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:29:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1cbcf134857462abfc8efede83d1193 : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f12c7de62324ed083ff4dc6413a86b3 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65f3867f24e4baa96d41617f5b39b11 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd1bd04492ee4caf99cc0c55387c362f : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9721e70fda674c75bbf6be899541b5ba : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cccaf84c6b442789d4fbdc2b31c25e7 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14345883e0e546178c0ba9c4fff2281c : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/338601e6b652469c9595f7fa81810628 : FAILURE in 5m 56s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab4213978d2047b5a84c1d12d9bb0dba : SUCCESS in 27m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/89342cc449d341f69aa000d0fce49402 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/820080f2ff4e4eed8136aee9eeb00227 : FAILURE in 55m 37s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4ee8c85fea0d4abab0b82d89438ea09d : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c7366e9f6a384a2bb4ac113351270e13 : FAILURE in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fe915e12190e0ace7a318a97bec59fa0f39c28f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 09:23:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04ef648108ed0ce3dbb22fd40f083fa6a22ccbf7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-29 15:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"515c56e6359e71b5d1947d22cae425a9e0265cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:12:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ba68378392f5514ef06c95130f89d9b622008f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:14:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3fc8ad29319b48a3b5f88aeb67a95820 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb540e3cc664243ac3ee49ca8628d97 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/843d88ff7b604a7891f4d841908e4882 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c05297d127a48adbdfa607bc7cfe42d : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0745fe0dcc04fa1b4d3a58b944a0dc1 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37ad78bfba274f388a08b9cf94bbf065 : SUCCESS in 5m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7187e63b73944e578a858525805da1d9 : SUCCESS in 17m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5a59ba1f84a0418d9b01e0f31467963c : SUCCESS in 1h 07m 45s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/457bdd98eeda4023b4a5b911d3c9597b : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5553fe4fa595af6f359a4613a04104ea4bce18d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:14:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83447ce1b8fe6cb563de9592ac67f02aa38de9df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:15:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a902c4184d1a44628699fa508768b657 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e0c3300e82f46a5a7c07889f2f1b578 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79b43f951b4d4498b7506fef37b81b493ef9729c","revisions":{"79b43f951b4d4498b7506fef37b81b493ef9729c":{"kind":"REWORK","_number":1,"created":"2020-07-22 17:12:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/94/742494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/742494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/742494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/742494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/742494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/742494/1"}}},"commit":{"parents":[{"commit":"36b8438297afee5d08945555dfe5c74a90fd35ee","subject":"Merge \"Don\u0027t disable SELinux if it is not installed\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36b8438297afee5d08945555dfe5c74a90fd35ee"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-07-22 17:12:59.000000000","tz":0},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nSets \u0027monasca_install_type: source\u0027 to remove need\nfor kolla-ansible var boilerplate.\n\nAlso use default Monasca parameters to configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\nStory: 2007597\nTask: 39587\n(cherry picked from commit 39308a45914cc452362b6f4442785ddc7e0c26c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79b43f951b4d4498b7506fef37b81b493ef9729c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79b43f951b4d4498b7506fef37b81b493ef9729c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
