)]}'
{"id":"openstack%2Fkayobe~742495","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","project":"openstack/kayobe","branch":"stable/train","topic":"monasca-defaults-stable/train","hashtags":[],"change_id":"I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","subject":"Define default variables for Monasca config","status":"MERGED","created":"2020-07-22 17:14:58.000000000","updated":"2020-08-03 15:41:52.000000000","submitted":"2020-08-03 15:40:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742495-1596469236796-7b2e085e","meta_rev_id":"e4c17483de806be82c5450116f21f60156f83907","_number":742495,"virtual_id_number":742495,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 15:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-29 21:40:16.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-29 15:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-07-23 09:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-29 15:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-07-22 17:14:58.000000000","_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 17:14:58.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-07-23 09:22:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-29 15:09:31.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-29 21:40:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-03 15:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb755e68e59dca6ee9d286d3d62fc7ad435bac27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:14:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8f1c38414fcdea718f49dc0f1ed9502d19b6610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:32:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af1512c0e66644dc883d3ee09ccf18f1 : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/521bf7d7c682419e8dc9300b9c932fb1 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0967d097cbe4b4385f27982899c5133 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41386bfb27fb40a6827eac279230448d : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7405d117d9cd4c699b0a9011824e1d20 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7af20a6b4b94af3806128deda5199b8 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/787d040a6bed4202a3bc3bf67b59a866 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50254b480b8946b5aacf82c5b9fdeec3 : SUCCESS in 8m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/812cb564f2684240ae4143ede7203519 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59d9418f18ae4837b163596862d4a230 : SUCCESS in 36m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/cbbd1e9d776741698cc2d315af74eb22 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9edae49b271c4323b5323ed02e3fc558 : SUCCESS in 1h 12m 42s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/77f6062b41714bde8a337c2da294f4ca : SUCCESS in 20m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb5241c87a834f0b9dbecfc9f4dd39f0 : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f31a3cfd3f204814a4d90692b118d475 : FAILURE in 32m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a4e94259548440cea02a20f417aaaf06 : SUCCESS in 25m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7bf422c6dd9d47a199fd59ad1e02f7c9 : SUCCESS in 14m 57s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0450128dabdb44c7b2ae44d184d49b75 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b4ada47e5974403343df1cbfc51b2485e87864ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 09:22:47.000000000","message":"Patch Set 1: Code-Review+2\n\nIt\u0027s not exactly backwards compatible, but does at least improve defaults.\n\nJob failure is a kolla-ansible issue on stein.","accounts_in_message":[],"_revision_number":1},{"id":"e648e561fbb288c8b3886d2ed35e1c5fd6978132","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-28 10:08:53.000000000","message":"Patch Set 1:\n\nWe see the Stein failure again in the upgrade (as you would hope):\n\n```\n\"msg\": \"Can not parse the inner module output: /opt/ansible/lib64/python2.7/site-packages/cryptography/__init__.py:39: CryptographyDeprecationWarning: Python 2 is no longer supported by the Python core team. Support for it is now deprecated in cryptog```","accounts_in_message":[],"_revision_number":1},{"id":"2e078ee27e0ebe739b4949f390a3ac6a698254b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-28 10:52:22.000000000","message":"Patch Set 1:\n\nrecheck\n\nnow fixed in kolla","accounts_in_message":[],"_revision_number":1},{"id":"803962732eb78c924fcf318cac99e98f18bf571a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 13:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b94273b1d352481bb7aed470cde22645 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae8871b6396b4841b4b3840fee5d6157 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9d709d82924d8bbf59c3cd63a246f5 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af54587ccb714777b1b114c790c22591 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09a59214166449409c63f6c380b0e58d : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24adefcbcc84181bba880e9775111e3 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88845343826b47479ad35bfffc7c42f6 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89822d2f34fa4088b6f97d9ebf0ab454 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c6512741d3d4545b9b79c0038a9fe7e : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f16885e3c934d1c8fb35334465b8ab8 : SUCCESS in 20m 39s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9a80c31b7c604823916eb073adc3e6f4 : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f6478685946d407f892393bd375f8817 : SUCCESS in 1h 11m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/92cc0a0e08914f1bb5c127b92b03b89c : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7665fb07dea0404091f9ceef20225c45 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/092cf1ff972646f7a974f8b9d029e86b : SUCCESS in 2h 03m 05s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/eedf4ac1688049d387373dc98beda72b : SUCCESS in 21m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/91c76429f01647a3937b7bddca25c208 : SUCCESS in 22m 25s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e7d69497fd69402b8dd7593d78648d50 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c671528bd480d8dc1b76f3c454a48e87e4ff9ebc","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-29 15:09:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d5b21d5a0ecd7b7f2ed5505ee262b442ac4fb6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec525fd6fb7e9b8b6bcef91389ebfeeb47ce19e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:52:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1cf827d55a474d04a8bc2e595560dcb8 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/650f5418934848e28b5f1a8dc1a9954b : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d94d105450e84257944cc1ff7b53badf : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8967ed641a0348bebfe537342b2c76ba : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afc8c8e240244fea5d1cab47a9de7c1 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1f67257d72b4f0db1b3a017fd4ad1fd : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/decbc1dafeb0434c9f7a0668668d7b69 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/553174067b8d4bb7a33d57ebebf055df : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac04cfd91fdd4ef8b12f0fa0ad898b2a : SUCCESS in 15m 09s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/80bd8340334742b7bdda93b1541a290f : SUCCESS in 1h 19m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5dd8b7b879c045eea746d329e0990684 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/7ca79830c85b444cacc1ad07e88243c1 : FAILURE in 15m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/44d43ad4329342a38b2f15d4f0269223 : FAILURE in 16m 41s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fe5e47c11aa54e10b1bad3579bf5c3be : SUCCESS in 1h 32m 48s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5f4f2d86564e47819c4d1391fb8520fa : SUCCESS in 19m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0c7e9f2a3a11446c980eac7bfec56784 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a804c863f9bb4b5f848ab2596bb138a1 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f59f7b593a6c372c32240311ecf1f3ec4518ef8d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-29 21:40:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"288b04ca2beaad681ae7ab74df09157901fda747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 23:43:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14e78d5d54c24f52ba68a5c576bf9e1e : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c4893295d02448bad088d4c4e3179af : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6e01f69f25d4a4c91f237cded128df8 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae80cf777f444c9ca79f823be91a9819 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b793420ce37247b9a1cf925873150327 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458ca89de53c472682fe3a113e9a1633 : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/318a7c9ca24c477a98b189de68fc4227 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9987766c0aae40ef9f7cc13986abbfff : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/02a2df1f57d54599897a0cb6ceff3f60 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c101c06203e94cb982af48a3b07a7eca : SUCCESS in 21m 42s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d4424d08055f461cb1775c6be3a5a144 : SUCCESS in 1h 28m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a4ce0f70b6024a3ebc81003fe66602ca : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/871bf64d85a04b76a5b5657eb8e05b9c : FAILURE in 14m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7e4517049bc4489b939159932018b9e8 : FAILURE in 16m 25s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0d79da4a69f743578691a79dc0d59a32 : SUCCESS in 1h 54m 03s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/71728a7742cf42a1b1229e1c2604481f : SUCCESS in 20m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8025367b1ee24142ac98a83ea4895551 : SUCCESS in 19m 39s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0a621a0ef7754defa5d31105fd3d7b93 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d6833933134fa3674af14232f083900471f6610d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-03 12:23:40.000000000","message":"Patch Set 1:\n\nrecheck\n\nCI fixed","accounts_in_message":[],"_revision_number":1},{"id":"c75eca9f5485bc6491b7b5d9ef36c39e73814729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:21:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db2fc60f2e2d4d05b336ef16a3a5bf77 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e7fd73f652949abb31d289eda961a64 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4503256b4f0743c78c3c584e998b283e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542f0491c0014414a2552035445ddf73 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/188f8df7e006429ab40103d254994305 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3693edc01c49c880267e2a039a5e91 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51309b2446ee4b208a0c1ff288dda644 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3616134e7fb42cb82c3d2d1bebc725c : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d3e4656557547e48f3bcc07500557b8 : SUCCESS in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97c61890520b4af595755f8a69fde6fe : SUCCESS in 15m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f0682cb88c2e427c9344f018cd3b847f : SUCCESS in 1h 15m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b1be6162941c48bc8c85a0e068829db2 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c3988190a7724c52b73c984c3ed3895e : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3005e77dcb1f46198d244b72aaf0a455 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/40e3dfc07ca243b58466354a134ec7d8 : SUCCESS in 1h 52m 07s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/b9185f79077a44669df0527890473d4d : SUCCESS in 15m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8f0bd5699e544ad8b8b5ebbef5d1f929 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c23782e09f634ab7af22e7e4c5aa324f : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"107ebe3c01176dcdea9b55773103fc664544531b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:21:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d99353ac47f778eaa24d927265009c40734573ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:40:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1db84420173d12235f1954778694c0de9b39363d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:40:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6b43136fb1f4da38098b4c68117db94 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa26cbaf9ffd4edfa712efe659a0948a : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09eb47638d54a419547283766c9ac93 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23495abb980448058f100aacb76b698a : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3a87adaa73e478eba346cd359cb040a : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800b39e1a9c24fe8a6819156be1aae99 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c1219db2a0b4951bd3e178f7e4bc806 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53b4ce9419914583927eede6dc52c1fa : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa9ac62b53c6414786235c6f0224054a : SUCCESS in 19m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6e5ea7e5bdca4c09b7f9d55edc3992a7 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ee2ed21ab3c8443898e84cef70acdb2c : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f6fdb66b920d42e18e1698c583f55e49 : SUCCESS in 15m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cb7a06063d13415fabaf2387729e09ea : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2f6a7dd5abcf4098ab65439d28f6463b : SUCCESS in 1h 14m 18s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/458838db9d164e0f81aeb482f491928b : SUCCESS in 21m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f7852c9469374dbebc19f1a3822709bd : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/99904b8e20e54e6d8d73304f53aeaffb : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"072e698478a93c3e27a3de8ff3e5ad972c4ce4c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:41:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe672eaa53d4eecb90bd0c35690030d : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/575202e6b14b4e7b857a2cb1adb069ab : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47a7616eff53a0e66b144aab2913990e7cbf0fe","revisions":{"b47a7616eff53a0e66b144aab2913990e7cbf0fe":{"kind":"REWORK","_number":1,"created":"2020-07-22 17:14:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/742495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/742495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/742495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/742495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/742495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/742495/1"}}},"commit":{"parents":[{"commit":"9d15c847f89ef4e04924a2c6d29445fb682d2169","subject":"Merge \"Fix condition of placement image build\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d15c847f89ef4e04924a2c6d29445fb682d2169"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-07-22 17:14:58.000000000","tz":0},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nSets \u0027monasca_install_type: source\u0027 to remove need\nfor kolla-ansible var boilerplate.\n\nAlso use default Monasca parameters to configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\nStory: 2007597\nTask: 39587\n(cherry picked from commit 39308a45914cc452362b6f4442785ddc7e0c26c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b47a7616eff53a0e66b144aab2913990e7cbf0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b47a7616eff53a0e66b144aab2913990e7cbf0fe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
