)]}'
{"id":"openstack%2Fkayobe~742496","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","project":"openstack/kayobe","branch":"stable/stein","topic":"monasca-defaults-stable/stein","hashtags":[],"change_id":"I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","subject":"Define default variables for Monasca config","status":"MERGED","created":"2020-07-22 17:15:29.000000000","updated":"2020-07-29 17:15:49.000000000","submitted":"2020-07-29 17:14:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742496-1596042849162-7dc4a3d4","meta_rev_id":"c54e90477feaf43213f6354c7cba53e571147091","_number":742496,"virtual_id_number":742496,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 17:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-29 15:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-07-23 09:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-29 15:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-07-22 17:15:29.000000000","_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 17:15:29.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-07-23 09:23:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-29 15:10:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-29 17:14:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"792c20bed02c6e28ea3a2d69ed18f52886987429","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:15:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"72357cae3b34a74176e379089add75465493f888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:27:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36595fff41ef4940b427c7503436d980 : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46d60a40c37a4982a3cf29201e82ec43 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69fc9c4ba833464e8038eef4d35d6001 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c93a2884dc40ed9a43a251d0bb79e7 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8411cde9d9894bd79b189d6754176f2b : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabc59ed216945a785ebb55dfe62c443 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1670c563a384f27b6195d137abba613 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0556b701ccae450b90055f1a5ec0baba : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d345fec081144f10826c439809bddec8 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/786aa2bc93a145f4acc2dd80e2532898 : SUCCESS in 29m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e559cefcb3564a5d97ede772393bf907 : FAILURE in 35m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b75f04d8c0424e58b17e251db3d7778d : FAILURE in 1h 00m 45s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bbf1f06422434254a2f1e1286e784b98 : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ca67f3ca7d064e69abd43d9b9e3c8b1b : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1d632f4d6489303349c36a33c24627af8e368dad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 09:23:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4e268c5cf52e9011d7e01b583b89a8f06c9dfdc","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-28 10:01:11.000000000","message":"Patch Set 1:\n\nHas anyone looked at the failure here? \n\n```\nTASK [mariadb : Creating haproxy mysql user] ***********************************\nfatal: [controller0]: FAILED! \u003d\u003e {\"changed\": false, \"msg\": \"Can not parse the inner module output: /opt/ansible/lib64/python2.7/site-packages/cryptography/__init__.py:39: CryptographyDeprecationWarning: Python 2 is no longer supported by the Python core team. Support for it is now deprecated in cryptography, and will be removed in a future release.\\n  CryptographyDeprecationWarning,\\nlocalhost | SUCCESS \u003d\u003e {\\n    \\\"changed\\\": true, \\n    \\\"user\\\": \\\"haproxy\\\"\\n}\\n\"}```","accounts_in_message":[],"_revision_number":1},{"id":"3173e4690f97b691be58fe53fdf18caacff6eaee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-28 10:54:38.000000000","message":"Patch Set 1:\n\nrecheck\n\nfixed in kolla","accounts_in_message":[],"_revision_number":1},{"id":"9de1bed17e8b7cdeb56786f7a8c991bcd8a88a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b02ff34b8fa4b15b4e91ea097c4ce26 : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e56b2c285b9d4f5e8af9e3bc73cd8df6 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efb3c63bbec146e9a4d62f3d59b4802b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef952cb290024ff5ad5d1c0e3bad9f54 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6884e7da997547629beb38b51ac3f4c4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9ca71664e84ca5ba471f692daea792 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4722555f7a84018b3619323ee06428f : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b8eb0585b14479abcc9794256931054 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bd183b127b246ae83a14e2e2dba17b4 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1f64109928d449d854c9deb023925cd : SUCCESS in 19m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4517eac14834435c85c862870016bf64 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4e8ef19243664775adeef2ef4f0d49d5 : SUCCESS in 1h 48m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/01857cc15c36452a961d0a2110ca1232 : SUCCESS in 13m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/99d01030861447b99309fc621dc8fe9e : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2c15ef3299a0baa642cabe199c18b8dcda46c516","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-29 15:10:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks for the recheck","accounts_in_message":[],"_revision_number":1},{"id":"6e225e93ca283c9dbd8ad1fe5c3d94ba9fd37456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8926359993665490befb34e603cb796069a283d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:14:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31331be5ee7e4355b627317ff1c3938e : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c39562368a4e8981c716e83f6a127c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57eaf63ffdf4f6fb6f15190c396710d : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b0a1011482b48cb8e5b5e6e4b0a1a47 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335ce16fe3e2439aaf3b7eee3fb621ee : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1008a86bc7ea4410b4f8aebec9158570 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d68da38ce9643eba27e716c9014b08b : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd010e7aba7740a49e3473f8b296b2cc : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/702ea4ef95154df8a7850f21203c162f : SUCCESS in 15m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ced680723292458caea5ccd6ffe410dd : SUCCESS in 55m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7c02cbeb01014a1e8d0c2cf6fc9aeadb : SUCCESS in 1h 37m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/13a3b618388e45b6ba8ba155b372a16b : SUCCESS in 19m 17s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e3425d8772264720b66515fda8094067 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1c3d9ad226183f48b52df1e3b2848456f78a8b92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:14:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e05a224b8bc110dae21573e82ea198f270d8df5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed37043749a344b9aff78786d7de8cdd : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d87b4aa7be74eb6afb8dca0361acc11 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7e4d37b45bba872169df1f02991cfbe114307ca","revisions":{"f7e4d37b45bba872169df1f02991cfbe114307ca":{"kind":"REWORK","_number":1,"created":"2020-07-22 17:15:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/742496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/742496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/742496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/742496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/742496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/742496/1"}}},"commit":{"parents":[{"commit":"399e9eacf743c7b7a38a10644418d51c226dfebc","subject":"Merge \"Don\u0027t disable SELinux if it is not installed\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/399e9eacf743c7b7a38a10644418d51c226dfebc"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-07-22 17:15:29.000000000","tz":0},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nSets \u0027monasca_install_type: source\u0027 to remove need\nfor kolla-ansible var boilerplate.\n\nAlso use default Monasca parameters to configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\nStory: 2007597\nTask: 39587\n(cherry picked from commit 39308a45914cc452362b6f4442785ddc7e0c26c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7e4d37b45bba872169df1f02991cfbe114307ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7e4d37b45bba872169df1f02991cfbe114307ca"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
