)]}'
{"id":"openstack%2Fkayobe~742669","triplet_id":"openstack%2Fkayobe~master~I9d7f0bf7daa6d1abe31de0e76a508820ed123c28","project":"openstack/kayobe","branch":"master","topic":"public-os-endpoint-type","hashtags":[],"change_id":"I9d7f0bf7daa6d1abe31de0e76a508820ed123c28","subject":"Set endpoint type variables to publicURL in public-openrc.sh","status":"MERGED","created":"2020-07-23 13:39:11.000000000","updated":"2020-07-23 17:47:22.000000000","submitted":"2020-07-23 17:44:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742669-1595526291735-6361157a","meta_rev_id":"3a5083078e56ad10f73d1fce81de7573068a58b7","_number":742669,"virtual_id_number":742669,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 17:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-23 13:48:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-07-23 13:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 13:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-23 13:48:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 13:46:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-23 13:48:56.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-23 13:56:16.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-07-23 17:44:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99a4a3037824a84aabb06942aeb8743215077aba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-23 13:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b533c5c431371387370082fb8edcfb778fc10f8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-23 13:43:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbbc7b34ce529c43b32d9bc7dce613683586a6d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 13:46:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"89cdd0b26f1777c0e9dafc70d06b6ab1997a7b28","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-23 13:48:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dfbe20220a9da3fa45ef5a210f2d6c003a5ff135","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-07-23 13:56:16.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Pierre!","accounts_in_message":[],"_revision_number":2},{"id":"5b76ed2c12abd1da7f2b83a2cf8f995c86dc97f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 15:56:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89b6b5dc4e964c8dbd4ab661a00109e7 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e5af53786d479c8bd04e804a9451e0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51cd347a69a34bc881dd521a6498f823 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ddbb7a6f23d493bb39b7c43559af635 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192da6b1a4794af3b1a366b7abee9327 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/650407eb0849449c813d82709ba30782 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53b74e1ff41f4e899b784befea2f44ba : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/675eb092f63a4ea98143892c897a1feb : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a62634fedd084530a0a0a7a0226743bc : SUCCESS in 37m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9ebdce248aa04742a1594546be743ea4 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/34bb2e50ca294b5293a7d080945914c6 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0bbd8036c2594cc492baba31455d6c52 : SUCCESS in 2h 11m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6837e01aad634a299992f7e12d606a71 : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8cd5b3664ed444798384db0e328eb3a : SUCCESS in 26m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0aa7b234ec994ffa9f3989afe64c1e46 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e4632679d6c809e558692cf6d8b3d0723c84a4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 15:57:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b58a5b73aaddd27b0732f4a4a2d6e5fe936959c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:44:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2c06178e14ae136f9c3588698585f0e81447af4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:44:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c128adbad5d341029e6c4cbdd5f25614 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09703f58ef74ca38df0b168c433b312 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81745b042a724d2a873f4e4cebeef42f : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b2b1a439054da286fbc1edf23f7236 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f97d1ccfa294c78a7e4237ff4fb0768 : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c53d4e66f904939b9ce68a20abbf536 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3144e7e206f74536bd9d01d9cc95c752 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e10ee18adc074b4d95bf316f2bc8c391 : SUCCESS in 41m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2a4e156a55b648f695df7accf51c0b97 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d4188445fc5b44e69c6f66aaa519001c : SUCCESS in 14m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/57b8f35f2c314eea8b1c5226c7cf5aae : SUCCESS in 1h 44m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5d89f0b0e3eb477dbf5de1723b43f225 : SUCCESS in 21m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c43eb1321c840e9a2af62cf486958c0 : SUCCESS in 27m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cff32c961a1242bb90a45168d373e406 : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1d66e8318b5ba1b574d71257f449b642a2e3e70f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:47:22.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95c836705344b448df998365b3d7bac : SUCCESS in 1m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30d3be5e2e9442dd89f2f55b0c7c1ffc : FAILURE in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"271bc72dae91cd5d3e0cbf37f9c9931bbaec13f7","revisions":{"d89f56fed36f6206acb1306c5aac634490f80061":{"kind":"REWORK","_number":1,"created":"2020-07-23 13:39:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/742669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/742669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/742669/1"}}},"commit":{"parents":[{"commit":"97373c0519c60b276eb91a5dbcadc5a266fd9a7e","subject":"Merge \"Fix network configuration of network hosts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97373c0519c60b276eb91a5dbcadc5a266fd9a7e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-23 13:37:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-23 13:37:54.000000000","tz":120},"subject":"Set OS_ENDPOINT_TYPE to publicURL in public-openrc.sh","message":"Set OS_ENDPOINT_TYPE to publicURL in public-openrc.sh\n\nChange-Id: I9d7f0bf7daa6d1abe31de0e76a508820ed123c28\nStory: 2007950\nTask: 40426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d89f56fed36f6206acb1306c5aac634490f80061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d89f56fed36f6206acb1306c5aac634490f80061"}]},"branch":"refs/heads/master"},"271bc72dae91cd5d3e0cbf37f9c9931bbaec13f7":{"kind":"REWORK","_number":2,"created":"2020-07-23 13:43:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/742669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/742669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/742669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/742669/2"}}},"commit":{"parents":[{"commit":"97373c0519c60b276eb91a5dbcadc5a266fd9a7e","subject":"Merge \"Fix network configuration of network hosts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97373c0519c60b276eb91a5dbcadc5a266fd9a7e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-23 13:43:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-23 13:43:31.000000000","tz":120},"subject":"Set endpoint type variables to publicURL in public-openrc.sh","message":"Set endpoint type variables to publicURL in public-openrc.sh\n\nChange-Id: I9d7f0bf7daa6d1abe31de0e76a508820ed123c28\nStory: 2007950\nTask: 40426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/271bc72dae91cd5d3e0cbf37f9c9931bbaec13f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/271bc72dae91cd5d3e0cbf37f9c9931bbaec13f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
