)]}'
{"id":"openstack%2Fkayobe~743608","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Idbdd494f9435d663564346fd0e0a053b656168c3","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Idbdd494f9435d663564346fd0e0a053b656168c3","subject":"Fix network configuration when no eth0 interface is present","status":"MERGED","created":"2020-07-28 18:06:16.000000000","updated":"2020-08-03 16:32:32.000000000","submitted":"2020-08-03 16:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743608-1596472265449-3eeecb9b","meta_rev_id":"4672f7a7305da705ae3cfb564813fedfb64199cd","_number":743608,"virtual_id_number":743608,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-08-03 12:23:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 16:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-29 15:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 18:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-29 15:11:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 18:09:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-29 15:11:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-08-03 16:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b72a98efce800dcec7c6eae599255e6a605dbfa9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-28 18:06:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"544d51bcd363cb1c72fc7e1771c6397dba982b46","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-28 18:09:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0ecf594c1e1d6191211d930c2bd0eaa5f2f2a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 20:18:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/226527eac29d4b3dab7d673a29c57a1d : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e14a110026e4430b8c9551c9b9a0185d : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9de48e187464354a2b99e7daddb1bdf : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d31142cf0e44729b14804c4d4dd35a8 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3113d42ab57f4d25ab05d11ce91b91a2 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b49411bec84284a02e6cfa5b27ac24 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9f8f682811c4b398eaf063943318ae1 : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/991ef7dd25c74e478c71fa39d37b8093 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55c9b610dd244e819a768d9fa7591d39 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa1493c346be443c9566160185386999 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0c239c942b884cfc805f16e4736c5663 : SUCCESS in 1h 14m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/77b82164daf34d9f833d3458077febdd : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/725deae0c6d04aff87f0264c3210a5a8 : SUCCESS in 11m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4b81d536d044bdfa900febea1bc5a2f : SUCCESS in 16m 22s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a5bf35975adb4817b39ff6ede844d6e0 : SUCCESS in 1h 58m 32s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/36412f6b82664a039450a54aa3249225 : SUCCESS in 24m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3e1e186f2f574c19b322575a88344ffc : SUCCESS in 18m 20s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ee2f455e66ac44748884c7632010c0dc : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":1},{"id":"54e4734bf0aeb69a7b6a86e254ee7b32e401515b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-29 15:11:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab7683822945c23f43c6e05d3737202d7ac6eb07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:12:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fe2710900035c9b9286d780cca55506e9a6edbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:36:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b215dadc38642738016ef94e8fffaec : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64b3b426f9564e88b3513180c200d6a8 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16cbcea4ba1d49939ad1170741e0006c : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bda81ebae3b9490a8efd83898a52d868 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b81e6eb2cb4583b09b7f822981a1db : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a7118459fb741dfa2af3d7c2af203fd : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37c280a412b24d0a888c49672ae6b591 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1f987001f114a81a7fe3360dbb023a4 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28fe072b1dfb4a9893cde5afe6e5c05b : SUCCESS in 20m 55s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6a2f7c88b6f1412f86e95f04c13edf91 : SUCCESS in 1h 20m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f4e4295afc694084aed65cb82e62514d : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/818f0a62bd304088a4e93ffdaed4fce0 : FAILURE in 14m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5a09c6c0e403404d919ca4e20aa279d3 : FAILURE in 15m 58s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c7dbbc9bcdec41f59274f3533e9e7b0b : SUCCESS in 1h 55m 59s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/53a2c24776094b73ac0168feb2dd119f : SUCCESS in 23m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f84352a548ee4db48e2fdda69d125ac7 : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bb4829234a6149ff89b4d7c8d8b9ce1c : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"fed999701db48d98bf39a9691e493d5890dec617","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-29 21:39:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b5c5fd97a9bdd275d1b4c7a5018a0cc1e7f367b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 23:43:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7574284e159443c1aab935bb733e9e9f : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b1139b5117f14176a8107b057230f337 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1545efb7888947be99b9e7a1f8d9bce7 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540b4b37986246459884fe96fa08a24e : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43a8254100af4377b5ee5a85c1b2f7c5 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b28ab67ac4f422f864dbc0138dca5f4 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28194d294ec457082b7cc1154da3a00 : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/218f3c8bb5db49cea34d6c6364591fc4 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36db8888c36c435695e74f79a7880500 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bba86659f9f647668235cb606403ddcd : SUCCESS in 19m 47s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e2c24058464846179e27d7e8822eccf7 : SUCCESS in 1h 12m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eb2dc3010114478eab71bca3fde22969 : SUCCESS in 1h 16m 40s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ba5693754d5a4cdc95cd03cc5223ed7c : FAILURE in 13m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6a2e1acdbdda4320b36b1dfe4e845d3d : FAILURE in 15m 21s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3de53cad7abd459da1802a497dcacb51 : SUCCESS in 1h 55m 26s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0e0e9dfe9193408a9797c13650787913 : SUCCESS in 20m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2830c726cdca4d2ca3397d35afd7602a : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/106717f691ef4b43bbfb1a812ab62057 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f6e7346f459c714f9a2b8c9c108a23cef81a65e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-03 12:23:48.000000000","message":"Patch Set 1:\n\nrecheck\n\nCI fixed","accounts_in_message":[],"_revision_number":1},{"id":"a02c473c69ce763be65228d8bfc0a01849ac492d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:29:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e75b6df46d7743269b095cc74a9de996 : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef6d8378d39d4fb7bd78bb7b0ca407d8 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf1aee01a0c48f994385bb1f7300667 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d17f8e7edd424d83cf0b54ebe68f93 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e56d09e1c7d8431ba94610878e937cb4 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1431df3f484481b6718468abd2be79 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/708f0f7016c44646a6fe350f1c56dfaf : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/881bf8f04e5b4ff08d17fe25993ca0be : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49a27229ff894e1eb5cf0b34c6793ec7 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7bfe1dfe4854607a484c77e1ef29f37 : SUCCESS in 18m 14s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/68fb1d3514a146c3a354802b8ba4cd2d : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1db53c9f474846d3bb665439f02f7c61 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5f5cb5c40d3d4858b5eb34311c4ba5fa : SUCCESS in 11m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f0a3b509cac74d39919fed2b9bbcc117 : SUCCESS in 17m 03s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a91b168920f14ce1a8402519887ea1d1 : SUCCESS in 1h 52m 40s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/d98afcd177d44b668c5a460fa5b28eff : SUCCESS in 20m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7f607f3ec13c4d559c9bb27c792c1008 : SUCCESS in 20m 16s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8a7b279445164c18b75f5a4fb998debf : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e8e547542b1cb561790d89ffe3fdff03fe3dc8c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:29:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"097fd5270f50d729943a7356b8b50dff397d5b14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d59acc0b158d4580a06b30ade5e2bff1 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e52177b7de4135b384fde517946238 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b82a8d6589c478b931389a37158e4b7 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/272fff60f4454a0b9fc0706881934197 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9d8cf03ca546708f4bfc0b8dd60270 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3343ba283ae247ab879d7ba185d39193 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0147fcb66e34496db9edaa77d5d4e5af : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8ea439d76264104b8fa081edf12ba07 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/138bbdfad233410d8dfe4fc3ecd989e1 : SUCCESS in 19m 52s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/53d951512dc84624831d681b50a1b478 : SUCCESS in 1h 21m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ec44bed5fdf34a2a95f22d925f9608d1 : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/94cb3c55eec743e8bf8e3abf0462d2a1 : SUCCESS in 13m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2a6158461bdf4f359d73c22e77ff0e87 : SUCCESS in 11m 57s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/10b055eece564dab87953ec0604bb781 : SUCCESS in 1h 51m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/032eef5f8eb348bc9dfa6554da08973b : SUCCESS in 22m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a72cf3b0f82547fc8c96e42087f2a7b6 : SUCCESS in 17m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6485b361fad74c54810ab119c8e9e984 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a13f39914e6cc339b7da4de24b939f95e8500eec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4672f7a7305da705ae3cfb564813fedfb64199cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:32:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37a4a26b7fb481eb3ad0c02c377b96e : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/090c5680c167479390ee412d5911c707 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"780fce2b78481a57b8c133a2e4bd69d8ac4b7e55","revisions":{"780fce2b78481a57b8c133a2e4bd69d8ac4b7e55":{"kind":"REWORK","_number":1,"created":"2020-07-28 18:06:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/743608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/743608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/743608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/743608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/743608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/743608/1"}}},"commit":{"parents":[{"commit":"0ce229866d8b0cc773c39eb9d70383d895a0cefc","subject":"Update seed VM image to CentOS 8.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ce229866d8b0cc773c39eb9d70383d895a0cefc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-10 13:06:01.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-28 18:06:08.000000000","tz":60},"subject":"Fix network configuration when no eth0 interface is present","message":"Fix network configuration when no eth0 interface is present\n\nChange-Id: Idbdd494f9435d663564346fd0e0a053b656168c3\nStory: 2007913\nTask: 40316\n(cherry picked from commit ca648622b1b2ca73989fe2011630aaeb3601f8c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/780fce2b78481a57b8c133a2e4bd69d8ac4b7e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/780fce2b78481a57b8c133a2e4bd69d8ac4b7e55"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
