)]}'
{"id":"openstack%2Fkayobe~743934","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5","subject":"Sync kolla-ansible inventory for Ussuri","status":"MERGED","created":"2020-07-30 09:23:22.000000000","updated":"2020-08-06 18:36:22.000000000","submitted":"2020-08-06 18:35:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743934-1596738911476-ea02c699","meta_rev_id":"9c4d40715de30b5815b63c1dbd899ba2b6487b0e","_number":743934,"virtual_id_number":743934,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 18:35:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-06 15:04:04.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-06 09:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-30 09:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-08-06 09:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 09:57:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-06 15:04:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-06 18:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"188debe3679585a8c90be647222d0ce5feaa928e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-30 09:23:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"51ecfb5dea4b9c2d162f3bac69a9b0408ce6c770","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-30 09:23:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0fc9f230ff6c4ea82701a7afdcfebf8ea5ef094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 11:49:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ca4227227404728a57e587d5f4f6b10 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa67a535b7f4b2ca76b9399d24efa0a : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fe7adaa535b4a20a43e1224c18f6c6c : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5677a3d20f1640d18b0516b261fd8af7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd81d3753574222ab387c20b154c291 : SUCCESS in 6m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f8092fa15fb4dec9e96477542b3ee67 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68b61ddb9b9f4d8289038efe0aff053c : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/230ea040e5db411daa728a603847af0f : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f244f82220cf4a0fb91942209b6988e3 : SUCCESS in 1h 07m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5461582522fa4621b69fcdf40198804b : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2c18b40be80347618a7241a79e1767f9 : SUCCESS in 2h 17m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/327acd08d26d457786d1d0268925a163 : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b917a63c911e4e1bb771ee9c6779bf83 : SUCCESS in 26m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e0072484c8384e6db4fbfcf0b6fbd080 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f0229ea7f8f0f3e0400bba4f942d89120e888d3e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-06 09:57:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed9e4910c86e3304da739cdac7fdec3f0e3bc373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:57:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d515e5c7f046426ad29db9d342768ebba5c04787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 12:01:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20842d7113ea4422b60adbfabfd81321 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d103af925324143b0652b870b662a40 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8055fac6ed384142a39aa74357f8ffc1 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd62a7e9b494a75af3f990efe133592 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/118321da097a4ebdb97014e785b40b20 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/302641423de44d2ead702cdbeda018ba : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca8c837c4ed84d35be8a6744e1c37a4d : SUCCESS in 19m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ca9c507f2e564cd3a16948c26ec46417 : SUCCESS in 1h 12m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6c80a1b9fae24b0f89d183f19148e896 : SUCCESS in 18m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ee9f4a57d3c84b9693edce4b9fc8029a : SUCCESS in 1h 58m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/269ba341de4c460484160724ba3ce0d9 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a97f61ace22a433582833eb75b782320 : SUCCESS in 18m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d334ba2714e44773892b175b1f29073e : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"704e26d9169bb3314599492cfee3aa53c3d02d09","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-06 12:06:34.000000000","message":"Patch Set 1:\n\ninternal error: qemu unexpectedly closed the monitor: 2020-08-06 10:07:38.364+0000: Domain id\u003d1 is tainted: host-cpu\\n2020-08-06T10:07:38.438496Z qemu-kvm: error: failed to set MSR 0xe1 to 0x0\\nqemu-kvm: /builddir/build/BUILD/qemu-2.12.0/target/i386/kvm.c:2119: kvm_buf_set_msrs: Assertion `ret \u003d\u003d cpu-\u003ekvm_msr_buf-\u003enmsrs\u0027 failed.\"\n\nI\u0027ve seen that somewhere...","accounts_in_message":[],"_revision_number":1},{"id":"59856faac07fb7b87d7f308c25600b13592bb700","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-06 12:24:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23b3364bf9938a0efa781c68c0a534a8862014e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:43:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8749f515a544e52b956c9a4011c30bc : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f155a45397ff4415a4c1cdad45f1e23f : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d621b9f6fc4a948165ee8f822136fd : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ff171d08b6b45b0a30095135e2bdc8a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29735814ffb344ad89869847effdf138 : SUCCESS in 10m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66f62cdecb024f6d96102f0744c34b31 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/732e8bee5d4a46f19a4cbcd9a92736ba : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0163912c18eb4e5481bd540e7bffaa14 : SUCCESS in 21m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2de7e9b7c490434eb61581897b9e108b : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ee88b78977bf403cb7a3a61330cf2e1f : SUCCESS in 19m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f34609252244fda9da260af7a62562b : TIMED_OUT in 2h 34m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b2ee6a4d694445dabcedced5328d2e4e : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/14b11820f7b248e7af9222612912b3e2 : SUCCESS in 28m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3925a78a7f08418ba59dac4c0ac2574d : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5056d9277f400543005c8a41ffee7ecd4bccc8ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-06 15:04:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35a32ffd12c16d3996ee751bb75b80d25b98b87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:32:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a379b68eaa49178b0e1c94ff728c1c : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270890a2308e47448d9b1eba110af536 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dfb3e9ee86c49b4b408e503fdd19689 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2c7f7cf75064e7a9c7c65bba06ea986 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9545fed320c745879d62dd860aac90d4 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ec33bfa16994353a689ce34fdf1451b : SUCCESS in 8m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1df6557d3dac452dba6341cbd10abdb3 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bdb06b9ddce4abb8dec5cd19cb15b42 : SUCCESS in 19m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a97fe7317fa44b169f1406331058f25d : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c246d1dc31014bc28826f7d41e780d68 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4cc02ac5bb7a43c48635bab6639db689 : SUCCESS in 1h 22m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/40f4eef142e54d518bfbfc386332c697 : SUCCESS in 19m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/66ea3ae9e85c4140a8b2494c0a9662fd : SUCCESS in 21m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9d0fdff1280247aa8517c0a58954521f : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"54140b15dea8fd0a72e9f8a75f1fce4d2b48fe4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:32:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c93b39eb464bf386f9d5509967ecf1b5d62bb77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:35:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afdcb1ac77f4983b1b8a2dcb9e6a3df : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d322487f694bf69c68f2c4a794cc7c : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13d18ee10c854416b68fe5927bd52712 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d4bd0dd35541d8b84d12666c3a56fd : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/118a8e826c6e424aae8b971d566439d4 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ece1e261b2624e07b3404144ed19c6ae : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c47abe68526140ff94b36e73c2549b07 : SUCCESS in 19m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1f1da8bd336547aea26a7b3622a8402b : SUCCESS in 1h 11m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7800a858c1924052b3bff75f209b269f : SUCCESS in 28m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3706421cee7a43fbbf6ebe5f720b368e : SUCCESS in 1h 57m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6a945fc0c8664746b636e577cbe1d6d1 : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/47a5f34e26cc40a7bcc4411297be4d55 : SUCCESS in 45m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12939ddd9ca541659d9216bebaaeccc1 : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"86406a73d4e1350a5f8efdbd2009c6ae08b50ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:35:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c4d40715de30b5815b63c1dbd899ba2b6487b0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:36:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66041e9dae64ca6b75efe53fb0a6cc1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c65cbbcdde92e03726f443b09ee187ece851859","revisions":{"6c65cbbcdde92e03726f443b09ee187ece851859":{"kind":"REWORK","_number":1,"created":"2020-07-30 09:23:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/743934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/743934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/743934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/743934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/743934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/743934/1"}}},"commit":{"parents":[{"commit":"eb403139839c69b84e303ddc6de6c53ed126c8d9","subject":"Merge \"Update hook link instructions to use relative path\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb403139839c69b84e303ddc6de6c53ed126c8d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-20 13:56:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-30 09:23:22.000000000","tz":0},"subject":"Sync kolla-ansible inventory for Ussuri","message":"Sync kolla-ansible inventory for Ussuri\n\nBackport: Ussuri\n\nA few services were added at the end of the release which we missed.\n\nChange-Id: Icc5d1e22a1e0f3b5a9cc23a7c756335b4d4a13c5\n(cherry picked from commit 936f393d6329982efbfbe876237a96f1a40c92b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c65cbbcdde92e03726f443b09ee187ece851859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c65cbbcdde92e03726f443b09ee187ece851859"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
