)]}'
{"id":"openstack%2Fkayobe~745569","triplet_id":"openstack%2Fkayobe~master~I443da1ae05fcca2d7d8dff7db563eeda37e9f502","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I443da1ae05fcca2d7d8dff7db563eeda37e9f502","subject":"Performance: avoid set_fact in Kolla Ansible host vars","status":"MERGED","created":"2020-08-10 16:20:48.000000000","updated":"2020-10-01 14:41:51.000000000","submitted":"2020-10-01 14:28:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745569-1601562528407-8f1aa1ae","meta_rev_id":"615dc602e7e85f342daef6cfb34aa99c2f2cbfeb","_number":745569,"virtual_id_number":745569,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 14:28:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-10-01 09:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-01 09:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-10-01 12:22:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 12:18:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-01 12:22:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-01 14:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f5c636f0c850016288b668336b957cd7dac0412","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-10 16:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd7c5db6cab7fc7b76b1ea90928a478466a58a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 18:28:03.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/790b70002aab44b0899e1151fa6ad3a7 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8a802d873a4c5bb5c923392198d865 : FAILURE in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f20697803545f2ae9fd289acad5d7f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34eaf11f3ad74c0d8316449c0711a328 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc47a44d40b34a0eac7091ab53e3b18c : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc9aa71be479462285ff336d28f3e499 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/314ceb1581b046d9ad9ac30b9e771ae4 : SUCCESS in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/029960ff341342a48877822d295e7b58 : SUCCESS in 16m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9da8dc9035eb4109b7add88e4cb9554f : FAILURE in 1h 01m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ef5fe0807d8e4b08993ab83a380bef1d : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef4b0682dbb44272857f39ce173a74d9 : FAILURE in 2h 01m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5505e0295db345fa989362b308ccf59e : SUCCESS in 18m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/081f43d0c7a94fd4bf798c9f546019ed : SUCCESS in 29m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1dfb766113c24a49b68589ac6d80f550 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9b64521d7a2a0a563710a964e268bb5475413946","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-12 12:26:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14cfd453d1083ab99334ada64faf9e8db1a695b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-12 12:27:17.000000000","message":"Patch Set 2:\n\nNeutron agents failing in a weird way, possibly an issue with yesterday\u0027s images.","accounts_in_message":[],"_revision_number":2},{"id":"8e4526af4f4aa73d94bebf440d062e7db51938c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 13:21:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a98ec5669a244b79bcb479377786586a : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2514cf77034224818d652720324330 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8fc3974cb854baf88d1194250a9bbcb : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c761eca09fb43ac93119a6a9d467469 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71adfe2f4b0248dcbe2dd6fe1b0abd1e : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4e99804b04e428bb8377ceea811d64f : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c6032e3f07a4a58a6a46dcf9a9cbdcf : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02a9b4b053ce4cb697157a292e6ce803 : SUCCESS in 20m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dca019a319504d79a57bccd43f86bb69 : FAILURE in 42m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/052eed1d0cf343e596a9b3ccc63eaee3 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/164d822884f143a799bae21bb76d3532 : FAILURE in 46m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d0eef6d41c34721a4554840cf7a3708 : SUCCESS in 16m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f8ad3d61b9814da0aa4bd7dbfd36df7f : SUCCESS in 25m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/93e89a3730ed4a0394d0079d64f065ac : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":2},{"id":"dc373f9830ea522911d5a5136b12716f6c3fdade","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-21 07:20:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d1c62f05f7694d09733803dcc152aa7bac5861c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:26:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d1d648252a24f5bb8fd877e86183253 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13736a6e70f41feb6596fd04eae58f6 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da12901ad8b04fde84638b2340eed3b9 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ee3edeaf8e4227bba1c1fce32945ba : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5562d632cf274735b9bd0c2f2f511839 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a759c9babf3c48ea9b5957e4e5eb5733 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ccdaf2399e3744b5b6770836b8d2bf2b : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bfa95e5fd4d402b84d2b95de7033475 : FAILURE in 8m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0e8068ebddbb4482aa2cf3b8e87eb684 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/026db678ed0d4aa7bd650942f604a12d : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/769d2e45f0054f1280b4a59f81321953 : SUCCESS in 1h 56m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3ab16df9ffeb41f995cf258ff2096789 : SUCCESS in 14m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e17fc27b1f664a06bf08018429a7dace : SUCCESS in 30m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3ef93633cddc441ba9d6e34baa91700f : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"308b2de33eee92aee7ed5c28a9a44c3867a11bf9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-22 16:06:23.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b0583c1caf60cb8b251116002288d5b57496ccfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 17:50:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46758129d4324405a2740fd3de44e52f : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9d1ad05d8943709ef2a38f92f00839 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba10626e195640a28d94db9e1c40403a : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36fe4165cda542d383d8aa3218f01981 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae30e28badb64c2d8a0a62a1e02e0654 : SUCCESS in 15m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0080f44ab9954b53843b921238bed92f : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3e5b151ed66406e87449603939f5c89 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b633a5522a64a1b8b870e15678c6e37 : SUCCESS in 21m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6a8193579a6f477db0ae769cc98ea3f4 : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a4eb9f6a4e414487b5941918bbfed8ed : SUCCESS in 16m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/091838db2f7945669cd34b7898c6ac9f : SUCCESS in 1h 39m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2bd6ea076a0b4e6084a1db37db3f367d : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0fa4956bee04a13acbae6b15e5b0d70 : SUCCESS in 25m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2fa7512601314680b3f7d67c77578619 : SUCCESS in 10m 06s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"b12b5f966a5084d9d12eba75c41776b6ba8ce057","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-24 05:52:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e5ae3bed33a4e52567d936f1a2c8ccf249fd7c37","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:16:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"90abe98e19044f195a63c559fe393df5eb84b8e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:17:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"da444fdb8e1d01dce718365a423c082bf631888f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:18:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"432d7d5adbe12a96edf28f812384d1d03e551ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:58:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb8172ad55d24a459248d77c206ffcad : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f993687348d04460adbe162968cf32e2 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91a8e259e16044cfa4ee10c1a30c7d0d : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a2fe79d44a4492f9f0f7d9716826221 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fa51afa3554cfdba2cd3b45fe93e91 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc7bb4d4c88c430f95466f83cb5b9f79 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/837e9fcf395c4b9b82afe66dceefa0d1 : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97f360d2552746528aec9440957ff2d2 : SUCCESS in 26m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3837231e0334428381f3fbf028385594 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d2b24104af8d45bc96c66b313806472f : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/773127b6858349beb769e2f256ac52a6 : SUCCESS in 1h 22m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cc4297b8d756426abb41ad5710a05b70 : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4ddf9ac0f1384719a09b3f9e3789872d : SUCCESS in 26m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8b6a276a252f456fb244b5d123789cfc : SUCCESS in 11m 34s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"4d1445fc3c7344e9110b26964474c0664bbc46db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-29 10:17:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"31176ebd5fe0a4c066205881ee59d61b82e598d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 10:17:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"45dee361a1eb426a4d60141c645257d4f427aab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 14:41:12.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff7b6301113497391e936a64c6aee74 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87484b730abe43ab84ae1154327a3260 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/360ecbc4878a4948a95a2a0657e9e293 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de03b6f2195f4acbbd7f9eb017949e82 : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b58521ce69d94bf69c6fa5c45aa06909 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86b7cc10361e490d9dabd45eac24dcc8 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/580409170d614e1ab524e06d9374b50a : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eb5a10e0af514a688c42f621ac4f5697 : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/442038707d244cb68a89403d145319a2 : SUCCESS in 16m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/45bbbb16120a4900b02baf4025bd5ea8 : FAILURE in 1h 49m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f558ecb01034560814f053f6d28d954 : SUCCESS in 18m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4bd54e9eafa546c68134a2db2b2e9e5f : SUCCESS in 26m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/52b83a4c50544414aa1982cf91fcaa8b : SUCCESS in 11m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"df3e6de386edd8a9b519a6f9cab63d5e614a0635","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-01 09:49:28.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8f1fee23c1232b2e5f7788d84dc02d14e1f9ba68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:57:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6e3466cc7c4a478e3954f8c1812a6c : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db210db8bcd4ebf836b8f877987816e : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c1bcb6a456466f8a2b0a7c89ecdd9c : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35bc36f2b4d34233a28aa229b762fb2a : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bed8d9f53849eabd4dd273b7d7b600 : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e05976a6d3dd4ceb9b3142b0ddbe9255 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/704c0230975448b7b0d161ed76bf1c0b : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c0a681ab6e04d6793a3c2f03ae8394f : SUCCESS in 20m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8df09230a81c4e718541d4b2ea6fc2ba : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6cf39758ecfd4f23b7f15edc447df5c4 : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/621da33bde65441d8de94aae96c0b6cd : SUCCESS in 2h 00m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5baecc47cd564960bd7ccce44b207d80 : SUCCESS in 21m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/823b7e5c9ea24985b4415ffc5c272006 : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b1b37a81c61c4cbf8e54033facf052a2 : SUCCESS in 11m 32s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"82821e8ddf828f7017c65f409beeb5e3ed0275c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-01 12:22:13.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ea77bd2d1ea3d81dc3b8d9e0abea4994ef8b2f43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:22:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0a6c4b7b465f380f6760eaef7a62020c7b89264c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 14:28:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d9c953c90567d486274b305280a394851126b2f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 14:28:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4a46fcef724f9b9742dd87b0731dea : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21890c0eca80429eb3c184cc4d2704f7 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f57a4a97b54461aad1e53b3346abcd0 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0725699aa25741a498eeb86dbc027d48 : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4a1dde2617c4a96a9e82e1936534c6c : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0cfc25194284757839675d481476b86 : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4b14f75fd7641eba41b4b178fdb8520 : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d35f8b2e4dfc407b805e68b14745e695 : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f83bb243d01b46c1a257e5eafa6b60fb : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f8efc200ab4943f098aa8ecc38fea6ad : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4fc604dcfc4749379f2ee8635a572a5e : SUCCESS in 1h 53m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4428c7b9559047fe88791f37b1989ad5 : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/49baa1cbe523444b9499f8589e120359 : SUCCESS in 29m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b709798c1ba744b9a10c45f9e63eb626 : SUCCESS in 9m 31s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"1f043a0aa94893452fcc41ec086d9363e3f4b0f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 14:30:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18093a1c18354737aaea8108f1002f27 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6},{"id":"615dc602e7e85f342daef6cfb34aa99c2f2cbfeb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 14:41:51.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit eb063dbcd957e9f1b36724081815f402923e2ef8","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5bf96da187c805b05254582012d93c8439c5574c","revisions":{"39bb4d6988c7d0cc2a4fe0025afb8a092580e6e9":{"kind":"REWORK","_number":1,"created":"2020-08-10 16:20:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/745569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/1"}}},"commit":{"parents":[{"commit":"f7b6e3f3ae9d2bbc0880802572d613a173f678e5","subject":"Performance: Parallelise Kolla Ansible host vars generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7b6e3f3ae9d2bbc0880802572d613a173f678e5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-10 16:18:44.000000000","tz":60},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39bb4d6988c7d0cc2a4fe0025afb8a092580e6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39bb4d6988c7d0cc2a4fe0025afb8a092580e6e9"}]},"branch":"refs/heads/master"},"b3755c80f5e0b11988c7f50497d4d3107dcf07b0":{"kind":"REWORK","_number":2,"created":"2020-08-12 12:26:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/745569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/2"}}},"commit":{"parents":[{"commit":"f7b6e3f3ae9d2bbc0880802572d613a173f678e5","subject":"Performance: Parallelise Kolla Ansible host vars generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7b6e3f3ae9d2bbc0880802572d613a173f678e5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-12 12:26:11.000000000","tz":60},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3755c80f5e0b11988c7f50497d4d3107dcf07b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3755c80f5e0b11988c7f50497d4d3107dcf07b0"}]},"branch":"refs/heads/master"},"b5d6c54142a9ff57da1bf3db68efcaa9ae9a9c19":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-22 16:06:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/745569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/3"}}},"commit":{"parents":[{"commit":"b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c","subject":"Performance: Parallelise Kolla Ansible host vars generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-22 16:06:04.000000000","tz":60},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5d6c54142a9ff57da1bf3db68efcaa9ae9a9c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5d6c54142a9ff57da1bf3db68efcaa9ae9a9c19"}]},"branch":"refs/heads/master"},"c1914e0109ce3da8ea26be2e0b16556b63b5bbb0":{"kind":"REWORK","_number":4,"created":"2020-09-24 12:16:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/745569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/4"}}},"commit":{"parents":[{"commit":"b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c","subject":"Performance: Parallelise Kolla Ansible host vars generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-24 12:16:56.000000000","tz":0},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1914e0109ce3da8ea26be2e0b16556b63b5bbb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1914e0109ce3da8ea26be2e0b16556b63b5bbb0"}]},"branch":"refs/heads/master"},"6a273781a39dfbe02a0497e9c30d1f20a304b71b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-24 12:17:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/745569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/5"}}},"commit":{"parents":[{"commit":"7c25fec6ca9526e7c7f12c05d45feb6063a88974","subject":"Merge \"Remove unneeded doc requirements in molecule tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c25fec6ca9526e7c7f12c05d45feb6063a88974"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-24 12:17:09.000000000","tz":0},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a273781a39dfbe02a0497e9c30d1f20a304b71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a273781a39dfbe02a0497e9c30d1f20a304b71b"}]},"branch":"refs/heads/master"},"5bf96da187c805b05254582012d93c8439c5574c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-01 09:49:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/69/745569/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/745569/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/745569/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/745569/6"}}},"commit":{"parents":[{"commit":"bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1","subject":"CI: workaround overcloud upgrade issue with ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 18:20:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-01 09:49:28.000000000","tz":0},"subject":"Performance: avoid set_fact in Kolla Ansible host vars","message":"Performance: avoid set_fact in Kolla Ansible host vars\n\nWhen generating host variable files for Kolla Ansible, we have some\nheavy usage of set_fact to set variables for network interfaces, in a\nplay targeted at all hosts. There are also tasks using the fail action\nplugin to perform verification.\n\nAt scale this has a significant impact, due to the number of tasks\nexecuted against all hosts. These tasks are executed at the beginning of\nmany commands, so the scope is broad.  There is also some tricky logic\ninvolved, which is difficult to express in Ansible/Jinja.\n\nThis change replaces the use of set_fact with a custom Ansible action\nplugin. The plugin executes locally on the Ansible controller, and\nreturns a dict of Ansible facts to set for each host. The plugin is\nexecuted once for each overcloud host, and returns all relevant facts.\nThe plugin also performs verification. Extraction into a Python module\nallows for unit testing.\n\nThis has been shown to have a significant improvement on execution time,\nparticularly as the number of hosts reaches 100 or more.\n\nStory: 2007993\nTask: 40641\n\nChange-Id: I443da1ae05fcca2d7d8dff7db563eeda37e9f502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bf96da187c805b05254582012d93c8439c5574c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bf96da187c805b05254582012d93c8439c5574c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
