)]}'
{"id":"openstack%2Fkayobe~746138","triplet_id":"openstack%2Fkayobe~master~I3de408d384141eee1a088d222dca7d583251a297","project":"openstack/kayobe","branch":"master","topic":"story/2008011","hashtags":[],"change_id":"I3de408d384141eee1a088d222dca7d583251a297","subject":"Block Ansible 2.9.12","status":"MERGED","created":"2020-08-13 13:46:50.000000000","updated":"2020-08-16 00:24:38.000000000","submitted":"2020-08-16 00:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746138-1597537415255-c5357c5d","meta_rev_id":"1c67b3e0d89d1e22de46867216ca3b7a778066a0","_number":746138,"virtual_id_number":746138,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-16 00:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-14 14:25:40.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-08-13 20:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-08-13 20:30:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 15:27:27.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-08-13 20:30:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-16 00:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"515c3253c5f1d88ffc25a684484f339fdcd43c47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-13 13:46:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bfee16ec80b3c33ef255229e532659b8c6c2a82","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-13 14:04:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6678d5f2cd7fbb20867aeb4acd5d2aead935e895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 14:53:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf7f0196aeb64216a038367dec0193c2 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6938ae6eff8e4b83afffa06c94cf4edb : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8c7821e23f42c6a2f665e05cb8b3d6 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7f106f4ada24f7fb6ab1f44be3a23b8 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd084714afe94e2d9343e494f581f4ca : SUCCESS in 18m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0afb435b063346c99543dfaab8f4600b : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0022948277c04e84b4dbff9e69840bf8 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17095684d7f44175a32638675f34cb33 : SUCCESS in 21m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4fba021bccca4ed3889b33d7e6be0f97 : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0b84ad7910a346df92ff058118b21044 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ba1beefd38714319b6dc8a5d124f78d0 : FAILURE in 46m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0fd0e3d34a9d44bba7b1ca15c42e024a : SUCCESS in 15m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2781bfbe1e2b426c8b75c2b1899e01be : SUCCESS in 26m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7c06240904e642fb8df65c39f28a4f69 : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"435533304cf5ed678d4a17ae2734d66208bcd7cc","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-08-13 15:27:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a33fe3e68ece4d6d1ab4ff3d4ad2b7eeeed2791","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-13 19:51:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit ab86a2d72783da997baf480670bca349fcff6102","accounts_in_message":[],"_revision_number":1},{"id":"8fb96699303e1b5cbab53308b9b58cf71cc9092d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-13 20:30:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"180c4b3c0584a02a1e959833e6f419b63d5571ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-13 20:30:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"870561dee02a5f8261cac3cd1f8b7b43f58b833d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 22:29:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e25178afe9e345ebb930241ce8ed2ca1 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6f6f3b9adb40dcb31b352e2779e3b2 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/401b753123bb4e4099b89a1bc3588368 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bb7c372de15481c87a1891f2f2b6671 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea5ffc5641a418ba2e604d828497df8 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02a3dcb1e5904da58808e85fcc71815b : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9339e5a4b174a5aaf74616887745c59 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fac9a27d2ab04b7099ae57f498c48e98 : SUCCESS in 19m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9479b5fcfab74195b92e6e463964149f : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a0690480c1d24fe38bfa49ccd9945224 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/44dfa2f2e6114fcfbff52b02109c47d6 : SUCCESS in 1h 58m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0c9ff7380d934294a8b17d4e0a5b407b : SUCCESS in 15m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9a8df52d4dc2432cae8319a49d473361 : SUCCESS in 25m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2ee7f987aa984188bd57bc5dee5ce68b : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e7c63b2e92da07970e8d4d1547de7612eb38b173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 23:54:58.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"74cf2c95104d74ef9707ac0d2d50a50daabab450","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-14 06:23:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3382368eacfde1630f93de27b5d95cabaec2b479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 08:42:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2340e28ad6f47c190cd018d1df8b7bc : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d9606bc31144ca9d5ade16c15be5b4 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fece49482bb428694fcef978bb3c8be : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/535f548f754b45a7a4e89c6338a4a605 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35489357f30432fac0f4809430e5667 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7cc814abd5b4d01bf37b48a28dde3b2 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5244ccdd4f94ce79481da4e8f4a07af : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b36d8bf0da2043d6838dc9f328848ec3 : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/efc5e565262c4cfb8bb3fe29fde1c513 : SUCCESS in 57m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2d770313f46b47d2a0660c3afec9358e : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2a1f722e81b34379870bbcc39a2da227 : SUCCESS in 1h 54m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a24e039c4d414c239dc671e6ae187fb9 : FAILURE in 14m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9c1a757403954d1d9424f8f6e2a057d6 : FAILURE in 17m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d200627cdf95423b98e3579ea6a5a9b6 : FAILURE in 8m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e0fa254f9afb27fa394a5d17986694447aef776e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-14 09:48:29.000000000","message":"Patch Set 2:\n\nrecheck\n\nkayobe-seed-vm-centos8 failed with \"internal error: process exited while connecting to monitor: 2020-08-14 06:50:28.824+0000: Domain id\u003d1 is tainted: host-cpu\\n2020-08-14T06:50:28.908410Z qemu-kvm: error: failed to set MSR 0xe1 to 0x0\\nqemu-kvm: /builddir/build/BUILD/qemu-2.12.0/target/i386/kvm.c:2119: kvm_buf_set_msrs: Assertion `ret \u003d\u003d cpu-\u003ekvm_msr_buf-\u003enmsrs\u0027 failed.\"\n\nTwo other jobs failed while running TASK [bifrost-ironic-install : Enable firewalld]","accounts_in_message":[],"_revision_number":2},{"id":"3d5deea4ff6119448efe0cba15a67e09e6c57a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:35:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab370796d65436d946f65b7be34eb4b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2696a2313014b2ab7a0de6cf80275e5 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1753e107c0d24b9b92c39a38103fd03e : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80aa65b9854b4dd193c4287fc635dfd1 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784a389f398941d5854ae8b3b7785400 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edd4a93668f54516b15adc67abf40c9b : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/138a91abff0c4fabac33604bf09f7fce : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12a752a8aa1f41f5bb2db558e01b9246 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2ac5f6b4df884d2d8b338479813e47a4 : SUCCESS in 58m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/20cf3d46577c4dc098254b44b0168812 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f85b2490d87e4ce89ca7cb2a6af0c466 : SUCCESS in 1h 45m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/036c13d2102c4d08b90c1f4f07334a98 : FAILURE in 14m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/959e8734a2284228ab14f14718c85aff : FAILURE in 25m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c98fc642007a40349f572cf6b58cf713 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":2},{"id":"699a22606e6c6435c9d794aa053159f879f9c023","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-14 14:25:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff42331759eddd3465ad35e6d3a1ed9145757e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:12:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ac92c7fafc24843af97230de06fabe1 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea0324ff6e34b7ea7132c01e7ae1ff8 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d28919ce9e4d76824761a4f7f03d59 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/087c8b45f8a1416cba160e74efe75b57 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133ede48492946d2bc3ad797920d61be : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bba44df8d4a4d04be560e997c180231 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdae6937fd1e45c9af5b69b2e81dba93 : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/777da5e50f64473c9e790aef34106c9a : SUCCESS in 18m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9335ad4389d04371bdf82049614e4653 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/57a53c8eed944eb1a2e3deab78e2c989 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2383eb5b2f254269ab9515deeb74bb18 : SUCCESS in 1h 43m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/33ed440257d04a249be0198527a71e6b : FAILURE in 12m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef8b661a3830479fb9157c87760d7ab7 : FAILURE in 23m 53s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/106d7ede3c644f01a4ab996607ea1e1b : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fe51a0f2b57c95319a4d71861d5dd4bb811eda6f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-15 18:10:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20dd4eeecfdeb036dc07e51e241194bf26f6a45f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 19:21:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bfb4b97fbe1499292a42f93651b2d82 : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2892c7a1a14beb8961c5bfc94b78e7 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9781eed741504fb686b629cdbeeb97a8 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9dbc2cf2679477db56c3798f5f74c9a : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1fb8d066c74778a52f649714df1789 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30ed4b048e6d4a12b1d16e388dc305b5 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e032c7d9a99046dea14cfeb6723c3455 : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c94a8c72f40846959394a4ea4a2038f5 : SUCCESS in 20m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/76ea020697d04b149fd3901adcdd5adb : SUCCESS in 53m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cd9887337b684397af130c4fe5365054 : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/06aa18e4a1224c9c8ec81bda61cfd040 : FAILURE in 16m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7b7e3f6f416b41509a507eae7435d910 : SUCCESS in 16m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c0fdfefe17ef49cea0ab7fc3a71875b8 : FAILURE in 15m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a33d2dab1b9a4198a38ee00020d43138 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2f7992dbc83d338d2c9e163e50872575ec4966ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-15 20:29:55.000000000","message":"Patch Set 2:\n\nrecheck\n\nMirror timeout","accounts_in_message":[],"_revision_number":2},{"id":"4f3e42330597d78b8ba5075c1abd6c501b700cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:29:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5af75441cb8c486385884589586c4c3d : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26b395be5b1417caa9a35f565f74b53 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3501ae97f7b1492ab63c986e6dd29a92 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7de85f4832c442be936e14bb4bc238fc : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c1674c7b2b43659d346b8d7f833270 : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee7b70bb81ae4e1ebb491ae3e824c210 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11eb8fabdfc94793b5cded5e6a121629 : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/839259b4c3224f829d918de0e936f21a : SUCCESS in 18m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7d199cfb9dbd444d9356699ce123d2f7 : SUCCESS in 1h 51m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3689f141a6084f8490cbe4593c7f22fd : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3f68b794b2b74345b03b953aa670709d : SUCCESS in 1h 54m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/127dc50ea01549bc9e4b71d11c23ae43 : SUCCESS in 15m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5eb85159924246b9b2243b96c791d5bd : SUCCESS in 22m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/511b316bc0184a1997f47ff483c4ea1c : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9ad1725f2b79b1080e22293ad7ef2a746acb3ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:29:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"97d7c071bacdbfc7b51db96b8c13ab4f73d08105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 00:23:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda8a3c27f664c2296d20d36c3661da8 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fee239ec337b4ebfa1d876e5fab93a28 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de63db148c6d4cab992188cbe2a5f3e7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6cf12e7f39a4b6b9be093b0af0ec464 : SUCCESS in 5m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70e72c287eab439d927c9f00ad968cf3 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f65da0fd4e44805ae4771942d69b801 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81f1f3688613425d96e397b07b9bb02f : SUCCESS in 18m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4f06d0b2c3ba47af870ac8c51b4e7724 : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6c08931fae5040ac9ce6fc0b1ff7f8a8 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4de3d02f746a4c519ed74d6178e5015e : SUCCESS in 1h 49m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d98bfa3f5e7d4970a107a433fe5fb951 : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a18ce15fef0d4a44af1a2d52bf7a4bef : SUCCESS in 24m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ee46fb5dd4394e46923d1d06e11f314b : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"47ce47127d72dc15e1c4fc1235d4f039a299e531","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 00:23:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2aad1102e4407678bff45a1db7b7d6f31f4e7bdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 00:24:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086a0b83d8f14aaab0448c389daafee5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6f08df6b14b2a575a597f01097c368d33d942dc","revisions":{"8ec08a7770b010218058cbcb89714774de4ac69f":{"kind":"REWORK","_number":1,"created":"2020-08-13 13:46:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/746138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/746138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/746138/1"}}},"commit":{"parents":[{"commit":"dce3545efb1f88b33679dbb4b5ae25c940250fe4","subject":"Merge \"Add support for custom Masakari configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dce3545efb1f88b33679dbb4b5ae25c940250fe4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-13 13:44:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-13 13:44:55.000000000","tz":120},"subject":"Block Ansible 2.9.12","message":"Block Ansible 2.9.12\n\nAnsible 2.9.12 changed default permissions of created files to 600 [1].\nThis breaks kayobe because kolla-ansible generates admin-openrc.sh owned\nby root:root with mode 0600, which kayobe fails to read to generate\npublic-openrc.sh.\n\nBlock only Ansible 2.9.12 as the next release should include a revert of\nthis change.\n\n[1] https://github.com/ansible/ansible/issues/71200\n\nChange-Id: I3de408d384141eee1a088d222dca7d583251a297\nStory: 2008011\nTask: 40661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ec08a7770b010218058cbcb89714774de4ac69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ec08a7770b010218058cbcb89714774de4ac69f"}]},"branch":"refs/heads/master"},"f6f08df6b14b2a575a597f01097c368d33d942dc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-13 20:30:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/746138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/746138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/746138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/746138/2"}}},"commit":{"parents":[{"commit":"dce3545efb1f88b33679dbb4b5ae25c940250fe4","subject":"Merge \"Add support for custom Masakari configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dce3545efb1f88b33679dbb4b5ae25c940250fe4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-13 13:44:55.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-13 20:30:30.000000000","tz":0},"subject":"Block Ansible 2.9.12","message":"Block Ansible 2.9.12\n\nAnsible 2.9.12 changed default permissions of created files to 600 [1].\nThis breaks kayobe because kolla-ansible generates admin-openrc.sh owned\nby root:root with mode 0600, which kayobe fails to read to generate\npublic-openrc.sh.\n\nBlock only Ansible 2.9.12 as the next release should include a revert of\nthis change.\n\n[1] https://github.com/ansible/ansible/issues/71200\n\nDepends-On: https://review.opendev.org/746220\n\nChange-Id: I3de408d384141eee1a088d222dca7d583251a297\nStory: 2008011\nTask: 40661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6f08df6b14b2a575a597f01097c368d33d942dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6f08df6b14b2a575a597f01097c368d33d942dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
