)]}'
{"id":"openstack%2Fkayobe~746200","triplet_id":"openstack%2Fkayobe~master~Ib51ac445cc787d7045d72a9ecec14434d432d9e7","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ib51ac445cc787d7045d72a9ecec14434d432d9e7","subject":"Performance: skip LVM configuration if no groups configured","status":"MERGED","created":"2020-08-13 16:39:11.000000000","updated":"2020-10-14 10:29:45.000000000","submitted":"2020-09-18 13:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746200-1600434376805-b51fe808","meta_rev_id":"e694910064276d28cd89826a8d626429e5ea54a4","_number":746200,"virtual_id_number":746200,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 13:06:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-09-18 11:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 07:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-09-18 11:04:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 07:09:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-18 11:04:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-18 13:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"556ea212ad8fadcb9c5acb12855d671262033960","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-13 16:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be8ef64ea92ad32a7efa9da4769c94047d9cfffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 17:32:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60756f8cd8f6467aa2a261aa7c72259c : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d71d5e5701d4f509b71ca0e2cd386ce : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c597a4cc738948c4953c723654c78807 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b401fccb3944a2ebdcec34cc4893ae2 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c1710259a3459b85c90a4f16c039bb : SUCCESS in 8m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2efc39bc6d1411b903c8ee27f12fd3b : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/29bf0a36d2d1464d9e01d535d27dee90 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c1ea3717cd5491cbab4908d250843c4 : SUCCESS in 23m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/61fb6dc477cc4d9b8cb30f9e721b4a9f : FAILURE in 33m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/721028900ab44eed80809a7b337194e9 : SUCCESS in 11m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8c2c0986cf58450e90ece88c03af610f : FAILURE in 48m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fdbcab76a766480294c2a5a1d93b5be1 : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5044ef9676b44941a080b8552e4e166a : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e153bebc870e42aba9855f74acf74c5b : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"78519be5765938f981482e9974b1acb2dcdf9610","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-21 07:20:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"992bbf3a2f349644f68417bcc2b3ac50cd88e02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:15:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d231fbf4ddc433f87b45a4cecb25e7d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944916bbc1d846d5b2316ae7ea40f0be : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/712e8a46ecaf4f32ba4cb2e4db683b07 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ec2b3359b943fdb66c9628d0e4b817 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7d10e83f9d4eccbd14119a37cb3937 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f81626c621e4030beba648905f3064a : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73ec11191de14c83b13f7d778939cd32 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f93924bf0d7b46a2a6b2706ad2ec5591 : FAILURE in 6m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/457bc639df16489693bc24622507c936 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d6893f657a3644dca939b0eb7b935221 : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8062fe240bd24cef8fee6baffa10ecd1 : SUCCESS in 1h 45m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e7a7044e0f584bdaa107a13965e25988 : SUCCESS in 19m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7e22b6f1bf374be3a04715941d7345d9 : SUCCESS in 28m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/50e4b8017d7744a19673df3072caba46 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ee95ff9ed1e4e37db5420fb3dfcad873e8a00b59","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-16 13:22:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e16f84fb75533012b4f14f3d823acb64f60165c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:17:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1612316f410a445e8661d5df0ad362b1 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029714108af743d2b091370f09a62b38 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b6be27af1614029879056335d465dc7 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff35cb2d33e0474199e9e7c44aeb7763 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcdd157c6ac4524b4987c9bf50a0002 : SUCCESS in 16m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/014b6689dbd7476eb7e1870dda97ba28 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f50e3206b164409a9fbea4c26e383bad : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/416f89c2ee1448a18619a5ff9200434f : SUCCESS in 22m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1da138e5410442c8b33878fcd84467e0 : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f86236da1b9c42539c5c88eb8f1153f6 : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/25b0946e6df24275a6a2d832ba845b69 : SUCCESS in 1h 50m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b1f900c203634eaba5c4df6f1454091b : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cccc2047f2c74e828bfe8814e4deebca : SUCCESS in 30m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/798abe107508461180dd78d9ee242b5d : SUCCESS in 11m 47s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"70b2e1dc07043c51ed6937f61245b98b75a8ec47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-18 07:09:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"183998f5233f79949535dd0271f3ef4369efc488","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-18 11:04:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e815cdb4292c64edb7013e535635b4f7f0bb2778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:04:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b98cc0a81ac36c7a14ca29ff9917cb12be63f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:06:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f196d2572bf4e40bfb66bdca6348702 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfa74b7494a540b9b9c23feb3c146062 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d7047a41ff644ecb057350bb315728f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a429bf1bab644182a000ec140d0c51d7 : SUCCESS in 12m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b96a736a536f40b1b6b0dfc698353b76 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a5273f6118c47c4987162696216f00b : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5af5274bd3f42dba82ca4fb742a2480 : SUCCESS in 17m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3bb4d315b212495e92037c1cc9190287 : SUCCESS in 56m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fa25589287a54ff3a77c688ca00c416e : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/870b912273eb43ebbdb4deae80b96b06 : SUCCESS in 1h 55m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5b5fb8409bca4edb8c67f887aac8d67a : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3484b343d2a140d6899a328ab96a7283 : SUCCESS in 28m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b467ad06f4fd49a0b42f949d535d6bec : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"4bda5f1cd2218c3a26df6e86fa64299882e4624c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:06:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea3cfcf14597ba7ac0d92f63c40473d50aaa6b12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:07:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce07ee7b2f64e828ae431c33898500e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c6ae1d8441dbde25b1445ceda71636f949573332","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:29:20.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7873dd85d2077f35c22d8127757d986e1c1ab79c","accounts_in_message":[],"_revision_number":2},{"id":"2e3f85c173fb0c13afc6bc6c7ba123c009dff98b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:29:33.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5fd45d67dea05e0ba3d4800b1d7b351b8b365565","accounts_in_message":[],"_revision_number":2},{"id":"e694910064276d28cd89826a8d626429e5ea54a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-14 10:29:45.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d143981542da846e8dec9abff985f018b2f4f0b3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ff5e6b10011875d1c29d16a8b39498adc0145e0","revisions":{"8fd82b6fc8554ff87c859c2db9a227687d8c03d5":{"kind":"REWORK","_number":1,"created":"2020-08-13 16:39:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/746200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/746200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/746200/1"}}},"commit":{"parents":[{"commit":"fd53039a0044509da8581c5c82021f5f0aa22601","subject":"Merge \"Sync kolla-ansible inventory for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd53039a0044509da8581c5c82021f5f0aa22601"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:17:38.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-13 16:36:01.000000000","tz":60},"subject":"Performance: skip LVM configuration if no groups configured","message":"Performance: skip LVM configuration if no groups configured\n\nCurrently we import the manage-lvm role with a condition that skips\ntasks when no LVM groups are configured. At large scale this is\ninefficient.\n\nThis change improves the performance for hosts where there are no LVM\ngroups configured, by using include_role rather than a static import.\n\nStory: 2007993\nTask: 40664\n\nChange-Id: Ib51ac445cc787d7045d72a9ecec14434d432d9e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fd82b6fc8554ff87c859c2db9a227687d8c03d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fd82b6fc8554ff87c859c2db9a227687d8c03d5"}]},"branch":"refs/heads/master"},"5ff5e6b10011875d1c29d16a8b39498adc0145e0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-16 13:22:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/746200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/746200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/746200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/746200/2"}}},"commit":{"parents":[{"commit":"84f98f4c04b856804630a930e777f76a88c60e2b","subject":"Merge \"docs: Move getting started to a separate page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84f98f4c04b856804630a930e777f76a88c60e2b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:17:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-16 13:22:04.000000000","tz":0},"subject":"Performance: skip LVM configuration if no groups configured","message":"Performance: skip LVM configuration if no groups configured\n\nCurrently we import the manage-lvm role with a condition that skips\ntasks when no LVM groups are configured. At large scale this is\ninefficient.\n\nThis change improves the performance for hosts where there are no LVM\ngroups configured, by using include_role rather than a static import.\n\nStory: 2007993\nTask: 40664\n\nChange-Id: Ib51ac445cc787d7045d72a9ecec14434d432d9e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ff5e6b10011875d1c29d16a8b39498adc0145e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ff5e6b10011875d1c29d16a8b39498adc0145e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
