)]}'
{"id":"openstack%2Fkayobe~746459","triplet_id":"openstack%2Fkayobe~master~I19ef5a46dccacc00db495993cc744938921996a1","project":"openstack/kayobe","branch":"master","topic":"aodh","hashtags":[],"change_id":"I19ef5a46dccacc00db495993cc744938921996a1","subject":"Add support for custom Aodh configuration","status":"MERGED","created":"2020-08-17 08:45:36.000000000","updated":"2020-08-22 00:48:47.000000000","submitted":"2020-08-22 00:46:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746459-1598057217625-4284198b","meta_rev_id":"5a6deebd52b4457403604f054a48a1287cb86c85","_number":746459,"virtual_id_number":746459,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-22 00:46:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-08-21 20:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-08-21 20:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-08-21 20:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 21:01:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-1,"date":"2020-08-21 20:54:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 13:59:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-19 15:15:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-19 15:16:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-08-22 00:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"badf8b652b719ff73b253605b157461772c80eb2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-17 08:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db6ab5b81e9e10bdb69080790710d2ff3aedc9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 10:06:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7733ed51329f471683e54bc42de08747 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38fe8885936a402fa71c03e8df8afbf9 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7033fddec4d846b493d6f866b9c994d8 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20ae9e678d89435786c2c98ad6c878fc : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c91e1d5e4c8417bb4951c392d25c653 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/104b8f95e9ca4c18a02699e9505841b3 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2589805dbdf4d86a3f8f27fc56f0292 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb1946565c4d448aa72a07aba1b62504 : SUCCESS in 6m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e140b88b0a1f4fbf925012b5dfd411dc : SUCCESS in 15m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bd02d2e7aaac4b059bdfd77ede2b6dfd : SUCCESS in 43m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4c7bb1b447ab439c92b4b363f6e76fb6 : SUCCESS in 11m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ae0d217f8b0249b09971985dcf611877 : SUCCESS in 1h 19m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a56add3e27c74d2ca271fcd753798638 : SUCCESS in 15m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/195c6be31aca45929447bfa6dfc36c41 : SUCCESS in 18m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/04d63adf5d444ef0abb9b4c55cf5ae77 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"213c97113d0a1d65eb99e9268095172fd37065f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-17 13:59:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf23ef94703d503750e7bf59c8c2e9fa27b058b9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 15:15:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c47543f97abc5c6f4d6380ab7553d9299561eabc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-08-19 15:16:43.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35fde39791be6ccdb56cb3d0f544e5ad8b652d9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:16:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35bd910a8becc35bd13c5a71f6c94fd94635667d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-08-19 15:17:14.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"becd65aa40492a0c61f5b434b5cbbf5fe239d344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 17:54:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e99ac255f5b645a6af65314284209071 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39eca5c1e00451bbda52a653e4c22dc : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487ba0b2ddab48d99829d45068c0ecc2 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62035809c344e9bbfd94b49627cb67b : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/990a9e55093a485299c868e556206c62 : SUCCESS in 8m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d3e8b90ada3499b811e1a60724ccd30 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4ba2bc2fe0a42fcbe1df0130397c888 : FAILURE in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1270948db66c4f83a67558f93f9427d2 : FAILURE in 6m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e2d6bc6cc0324964a0ea52b3673ba76f : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/85cfd09efce84a15b2e8b2444d7f65c9 : SUCCESS in 16m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c428490923054153b0a450b4816e8bcb : FAILURE in 2h 34m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5581575ef88f40b3a6be96d21fd8d16b : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a73a0d89ef5c44bca10cee0074e53eb0 : SUCCESS in 24m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a8909638d52044d2abc3d122408fdf83 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"02c5d3d036c00148dfc6e0aabf6388edf8411e3f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 20:54:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4df5487a09ee84dbcadc3e0526518b4eb480ae85","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 21:01:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f4ace623f072b4150309b29d596bdd973c1e922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 22:59:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8741ed7c7e854951a4815ccd5fbe5547 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777933a6f22f4909a0afcbfc67df2693 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09e0e6423b345489ea706c90b456642 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a28c2f1569f44a2a4014ed7ea496dfd : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9153c25c66024b20add9e24dc8874f5c : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b39e6480fa745eebeb762247dd11ffd : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f72017a69c0471fa920123437867d25 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6405442704744b439931f815581f14a6 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/24d60bf1863b43acaac2cf146ef7e84f : SUCCESS in 20m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a805b1a73e8e41aab33ddca489e9cfe8 : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/571dd43b48734b72997d81afa0fc2cb0 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3847a7cde9c54eb6805e728eaf7fc238 : SUCCESS in 1h 57m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e828cdb0ab1742a8b970db58ac5c051d : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/770784b0d3f041409d7763a06bd5c5d3 : SUCCESS in 27m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e9089b70c66d4c66b2ed5c1182c7b9b9 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":2},{"id":"896217f7d3c97b1e2d93c96c06ac5c5311c27f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 22:59:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fa3c775fffb3c8e7a53131c2e75d09222030d762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 00:46:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a35604afc539ee508766c4391a2d5d449ca76498","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 00:46:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97123433be248bbabd47cf1019f5024 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2d2d6a8bde48bc9f1b2e8892a6e4b3 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19ca01d87ad4d13a1811aee70f3d4f4 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2347e8aebff14a9ebd5dedf289e103ca : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e8b1f06757847a5bc3a2aae1fa86d32 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96fa5812b5d643a08ef05f1e0b33a8b8 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19d78a623dd14ee8a6cffbf5fbc0eb4a : SUCCESS in 7m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f290c54abb944fcaf801b3f19328300 : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0ef6f863e5854f1ba1c5ba9d9b8e5525 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/781cf2cdf69d4ab6b26a38e820c1c13a : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/72c2f76ebebd4b9db3fba671b75cd817 : SUCCESS in 1h 44m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ddc14c5915ac46c7bc4500ecd5f0ae47 : SUCCESS in 16m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a6f37a6331cd4a4da6aae3c005c67e4a : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0b38a2b1810242beb91a4a60d883c9f7 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"cfaffbb175d8b6da5876b5ff3ef38b3ef6deed02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 00:48:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82334d0a1165461ebb0a947e119df320 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/747a792f059c47d5aef62b32c25c3e97 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"878fe5728bbc74b3d42ab63c2c61de6c4bbd01a1","revisions":{"b5b74ece36b93ac47e38e27eece2046216eb5354":{"kind":"REWORK","_number":1,"created":"2020-08-17 08:45:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/746459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/746459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/746459/1"}}},"commit":{"parents":[{"commit":"f6f08df6b14b2a575a597f01097c368d33d942dc","subject":"Block Ansible 2.9.12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6f08df6b14b2a575a597f01097c368d33d942dc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-17 08:39:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-17 08:39:58.000000000","tz":120},"subject":"Add support for custom Aodh configuration","message":"Add support for custom Aodh configuration\n\nChange-Id: I19ef5a46dccacc00db495993cc744938921996a1\nStory: 2008019\nTask: 40671\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5b74ece36b93ac47e38e27eece2046216eb5354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5b74ece36b93ac47e38e27eece2046216eb5354"}]},"branch":"refs/heads/master"},"878fe5728bbc74b3d42ab63c2c61de6c4bbd01a1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-21 20:54:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/746459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/746459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/746459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/746459/2"}}},"commit":{"parents":[{"commit":"f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e","subject":"Fix ansible-lint after upgrade to 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-17 08:39:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 20:54:21.000000000","tz":0},"subject":"Add support for custom Aodh configuration","message":"Add support for custom Aodh configuration\n\nChange-Id: I19ef5a46dccacc00db495993cc744938921996a1\nStory: 2008019\nTask: 40671\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/878fe5728bbc74b3d42ab63c2c61de6c4bbd01a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/878fe5728bbc74b3d42ab63c2c61de6c4bbd01a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
