)]}'
{"id":"openstack%2Fkayobe~746468","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I9d7f0bf7daa6d1abe31de0e76a508820ed123c28","project":"openstack/kayobe","branch":"stable/train","topic":"story/2007950-stable/train","hashtags":[],"change_id":"I9d7f0bf7daa6d1abe31de0e76a508820ed123c28","subject":"Set endpoint type variables to publicURL in public-openrc.sh","status":"MERGED","created":"2020-08-17 09:30:41.000000000","updated":"2020-08-28 10:38:54.000000000","submitted":"2020-08-28 10:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746468-1598611020953-1d2acb16","meta_rev_id":"e19592bf9554b3c6b646ad3b660462937d03fe7f","_number":746468,"virtual_id_number":746468,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 10:37:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-28 08:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-08-17 13:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-08-28 08:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 13:58:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-28 08:33:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-28 10:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e79ba81bc33efd3ce60c4fe232d9f2b3f1f215af","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-17 09:30:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85092098297738e65f3395465e13c38811fea07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 11:13:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a1cd9b93c54980ba896e8947480e1f : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5e2dc33436b4f838a6779dbaa7e9c0c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044fc2d1ae7b4fd7ab04b855df6e91ea : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92eded466f454a269bb4f88750dc96fa : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fcddc763fe44c52acd4207b01f732c1 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87908847645143979d56436aebd4f1a3 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e6cfed6562542058fabe0088448a727 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f85c1028ed1942b8b32d91c2bc308ee8 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6d2881532b549f3adb8772c01432d55 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6393bbf3d9e94ce299e2699f88f78771 : SUCCESS in 21m 19s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ad58fef303aa4b018a847ba16de833d2 : SUCCESS in 1h 20m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/45a9f7e3e92741578ae8a514b043d76c : SUCCESS in 1h 20m 50s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5e347feb6d8f42848afd6474aeb9c151 : SUCCESS in 15m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/36845522b6684ff99162da395dccc306 : SUCCESS in 18m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9f799f10b2ba4d2eaf05c2f2611e7009 : SUCCESS in 1h 41m 44s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/91bdecf9d5aa4f0d910e4cdbb97dbd4d : SUCCESS in 19m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8c96cfbaa37b4eeebeab957dbd70178a : SUCCESS in 19m 08s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/76f77a04e2144886b6d5c56f0e39b2cb : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8ff25495c38da03b462be9f213ab09736b2a2e25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-17 13:58:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28a4dcb0562bd77fb435cb83af241fef72bef55e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 08:33:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84a594992cd36f134b5a8f6ed2841f3a3e5a6f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 08:34:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83772339690d9df64dfda7b50e69775a7c9b3b09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 10:37:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"69a0d2fed0f4eb3de66e37dd9cc3d6274ba4f5be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 10:37:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9fdb4f44de1c4c758e51b5e799e38c70 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c75b1a8c2d1b42ed9091a6e0d93e0ab2 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb80a07da6084661b6ea1fa0a7245b85 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3d00be5a652439f867332cc1752b8f7 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04c724c54cb4216abee1812980f5dca : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33574f259a5441b28abd6a36e26c8afc : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75801c9bf2a545c1bf0b898a5d7e269a : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d259e7bb85e457fbc3e919dd244e08b : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17123c7c368e452f978ad5a05b0dffa0 : SUCCESS in 19m 51s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/21cb586d665a4ed2b72cec3a56d3c5a6 : SUCCESS in 1h 35m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b596718fa0794752a4ff52429a843141 : SUCCESS in 1h 23m 46s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/e1d4bd8672d443d48df192f92f640b50 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/904cd9a5210a42668071fa6e2d73add4 : SUCCESS in 18m 58s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0a98a7751e474115badb375b1287e2d5 : SUCCESS in 1h 54m 57s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ebf0fec2510d40119e216643b7044a63 : SUCCESS in 21m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4cf7a54ff26a4347803d5fdfd53cfcee : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c7f6f9a4e43c4d1f8cc42b36c4a230e7 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3cddbd0f75d62d516e9b444fc42f18b3a8effc21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 10:38:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ff635124fb467db801e2ff5f129be8 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ba97777e2b744ca8719b2565eb257a2 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"673e1b9799f8d2d59ab09550a4ac267b671c2a59","revisions":{"673e1b9799f8d2d59ab09550a4ac267b671c2a59":{"kind":"REWORK","_number":1,"created":"2020-08-17 09:30:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/746468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/746468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/746468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/746468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/746468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/746468/1"}}},"commit":{"parents":[{"commit":"00461fc4f0b7f1e04bab07801deaf00bf3e95b20","subject":"Block Ansible 2.8.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00461fc4f0b7f1e04bab07801deaf00bf3e95b20"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-07-23 13:43:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-17 09:30:17.000000000","tz":120},"subject":"Set endpoint type variables to publicURL in public-openrc.sh","message":"Set endpoint type variables to publicURL in public-openrc.sh\n\nChange-Id: I9d7f0bf7daa6d1abe31de0e76a508820ed123c28\nStory: 2007950\nTask: 40426\n(cherry picked from commit 271bc72dae91cd5d3e0cbf37f9c9931bbaec13f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/673e1b9799f8d2d59ab09550a4ac267b671c2a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/673e1b9799f8d2d59ab09550a4ac267b671c2a59"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
