)]}'
{"id":"openstack%2Fkayobe~746955","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I849d9b09e973316308e574c447a78868e93a5420","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"I849d9b09e973316308e574c447a78868e93a5420","subject":"Skip broken ansible-lint","status":"MERGED","created":"2020-08-19 15:08:39.000000000","updated":"2020-08-21 11:34:14.000000000","submitted":"2020-08-21 11:27:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746955-1598009270658-145709ed","meta_rev_id":"88f7ad6a4568e1c1056fb03391ff035763219760","_number":746955,"virtual_id_number":746955,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 11:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-20 13:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-20 13:33:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 11:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f6fa6b090336fd385213670c53cb8415ec43570","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-19 15:08:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cf390b339b3aabf5e5e5107aebbd2f801c4b31c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 16:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe10d539033437596670b4b41b80960 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef37a962b716462e993a333b633df743 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3a26a2074d4c9289ade48d5a7f9729 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75b4b706c848481baed36eb7a656202a : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cf6a3987f44959b925d0731e3d84d3 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab0be9c3839e4d1b9336935dc2699c29 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e49d28aa724c44b4a713024539da3824 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9f254cc8b9d4be78dac28f3b45322ba : FAILURE in 6m 29s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a395687250224c2b9c613980dff8de55 : SUCCESS in 17m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f5b65369c2ba42ce919fc429059a4317 : SUCCESS in 53m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f0de274302974fe990b9f49c3af976dc : FAILURE in 12m 34s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/91002fd882d54bd689d3b1dabbc4bfae : SUCCESS in 22m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ab8a5d59a8f42afb5594961d4c1b24b : FAILURE in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a054545d408b879969b85c8b699750f8b972cb4c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-20 13:33:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c5ad1363c1cd3eacde3e220879b97d89769a36a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 13:34:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b59842b1169ac2b9ca9dbe3d3a0f0d5361a96897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 15:38:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37e95323d0014ed4ac2907f286586b70 : RETRY_LIMIT in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d63b3e6f354ca79eca4c285ea709a3 : RETRY_LIMIT in 2m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4e1e072f51b4f51af065422dcf483dc : RETRY_LIMIT in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2f30a9e0b34d94ac1d453381a22163 : RETRY_LIMIT in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96e34393c35c414ba8a5715f9a72327f : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa88d54cf1054156b57a4894f7b87c85 : RETRY_LIMIT in 2m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27699da7cd364fe5b404b833acb678df : RETRY_LIMIT in 2m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4d2ef41b07af490f8bb952442fab66a9 : SUCCESS in 1h 01m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e18f9c23279a463791c99bbd687d59d1 : FAILURE in 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8ef35c313ee971115de8fe080adc028bbcacf12b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 07:39:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6322abf7d76622e40b8f4808cf3f03668a8809e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:09:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15627a034b90406aadb243d89727a0d1 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b703825394d415d9dc70c245bed6412 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978c15e9e7394729970c419709dc8c42 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/708e40721bab4b47ad7a46bab696d389 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fde70b113646209f37b2b940f165e1 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51d50469ac9744dea5d42c61ccaef92b : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd8351bab5564d4fb31289e14114e6b8 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa66b90aa87f45f1b5c1596bc2434231 : FAILURE in 6m 11s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfb8903b89ed409eb068b78b011dcae8 : SUCCESS in 18m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/553d1d955b6648b6b4e9b36b2d042f36 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5c14a46a74dc4077988568fcf3a681cc : FAILURE in 8m 19s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/964a87d6f68444f0bfbfaf1ba07662ac : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6d4f66b26e1848dbba0f6fe679ef9cdb : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb7ae51d4a904ffa5ed875b4aec2506ed7a32aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:09:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69ba345e0eed9fedc426d03377c5f65dee1b6faf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 11:27:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c17104824993d8abd40fae69cb2439545a3663c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 11:27:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7941890b852d405380af78b06da90a8e : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbeff163de64a738d2aaac64c12eaa8 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8506aa0e643b4a6c957f9a3ed0ca1574 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5fd3c240704ec1ae2818c4fdb26553 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6460722421b54ac7a3e60670c3bb9183 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b88e9d5280374ccdab7ad8b84ffa9e1b : SUCCESS in 4m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b444c35ec844ccb93ae479bd2d0d0d1 : SUCCESS in 17m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f72500af7ad04d69802c1623a15c421a : SUCCESS in 1h 12m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f52ed2ab02a3431fb6b37d1ea8c85d3d : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"88f7ad6a4568e1c1056fb03391ff035763219760","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 11:34:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b7d69ccaf0451da03d5bc2379a2261 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/927c55708fb14e4596eecea697f5e3d8 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5de53a58f525503774326493265eb2132b3e3d53","revisions":{"5de53a58f525503774326493265eb2132b3e3d53":{"kind":"REWORK","_number":1,"created":"2020-08-19 15:08:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/746955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/746955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/746955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/746955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/746955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/746955/1"}}},"commit":{"parents":[{"commit":"79b43f951b4d4498b7506fef37b81b493ef9729c","subject":"Define default variables for Monasca config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79b43f951b4d4498b7506fef37b81b493ef9729c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-19 11:30:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-19 15:08:32.000000000","tz":120},"subject":"Skip broken ansible-lint","message":"Skip broken ansible-lint\n\nSee [1].\n\n[1] https://github.com/ansible/ansible-lint/issues/969\n\nChange-Id: I849d9b09e973316308e574c447a78868e93a5420\n(cherry picked from commit a33ba678df211c4d3f53c7ee9b066becd0e962b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5de53a58f525503774326493265eb2132b3e3d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5de53a58f525503774326493265eb2132b3e3d53"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
