)]}'
{"id":"openstack%2Fkayobe~746971","triplet_id":"openstack%2Fkayobe~master~I3de408d384141eee1a088d222dca7d583251a298","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I3de408d384141eee1a088d222dca7d583251a298","subject":"Remove XenAPI integration","status":"MERGED","created":"2020-08-19 16:00:31.000000000","updated":"2020-08-26 12:46:54.000000000","submitted":"2020-08-26 12:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746971-1598445899848-210ac784","meta_rev_id":"b53afab4449e4626962998d7dbc0a73ef13f056f","_number":746971,"virtual_id_number":746971,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 12:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2020-08-22 05:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2020-08-22 05:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-08-22 05:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-24 14:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-08-26 10:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-08-26 10:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 17:28:33.000000000","updated_by":{"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},"reviewer":{"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},"state":"REVIEWER"},{"updated":"2020-08-21 15:56:19.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-08-24 14:04:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-08-26 10:27:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-26 10:42:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-26 12:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bb6b54de2a2d79f90dfae3e7d240afff2e97293","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-19 16:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb27a861301dde21391d84462c0578410050e41","author":{"_account_id":20451,"name":"James","email":"James.o.benson@gmail.com","username":"JamesOBenson"},"date":"2020-08-19 17:28:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"14dc493576fdf2f0329a869b535a34278a36939e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 18:10:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/730a4049ee484e78b3da705a1cde962b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f45a0ca37e44e3a3addb28cc0f07da : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/159e7fe9ec9b4ebfb17e06afabfe68c5 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7322eb05e19842cfac3eec192befadb2 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdbda50bd0a4fa3894c9fbb7384d97d : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc7a92878f574ca9b3838d93eff8e01a : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/184f78ab58764ce9a16b3d6517b5d09c : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2071a009faad4f12aa9895ff918f79a7 : FAILURE in 6m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fbe35a73baf04fe681a981260c98b9c6 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5f4e07ffa19348aead45a73bc1a710ea : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1dd8c1f0a2a943b2b6d2065e288d867d : SUCCESS in 1h 59m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/79d49a2069c74550842a619170e67fa4 : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0756a30c3cef4270a1321d1e4cdf2e62 : SUCCESS in 29m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f8209c47e6ed4e3189be5f0e48e948c2 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6064a91cce8290b6bf50cbcfc87f72cb5efc7b20","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-20 01:32:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c9db9ad222a711086227978289c0a69d3a0a282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 03:32:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16372084a934483185d0ff4186604b63 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9424be969a470d837d5efb38d17e01 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff75e7729da4913bdb236fcc096c827 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71f4298d9f94d9ea926bc1cd5c254a4 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb48993e42694359ae8d2d2c604b4921 : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81eef380f243477fa136a91dc63b084c : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/958c970b1a5347e4ba58f2c57fac1eb1 : SUCCESS in 7m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73a58499cd8849fabb870f85e84b338c : FAILURE in 6m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5b373ea811e546eba828936d2a2c8e88 : SUCCESS in 1h 50m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a638dd8e3d824d05ba28aa8f7fd1ab65 : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9d3d4a4632c24f3fb68034083f85578e : SUCCESS in 1h 51m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f65fa33cf4b84b23a31422153777b18d : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/83924f5f3ca84621aeb22fe5143617eb : SUCCESS in 23m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3845a178eebe487b8b7858620f968554 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4852db894c29a0dad08c00bdd90f1b41ca34b791","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-20 14:49:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dfca4c5e8f01539fbd77ca65c9b0813c5166bfb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 17:24:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3af16cbad984c1fa908608b654a0b1e : RETRY_LIMIT in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd08d0f94b524cb8b7ad5cf7d98664fc : RETRY_LIMIT in 2m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e5b623b4e3345a5bf06b08e35559765 : RETRY_LIMIT in 2m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e5f1db4b0cf40a39adf11d24cb38f74 : RETRY_LIMIT in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a671f4414774a76882d2eaffac906a5 : RETRY_LIMIT in 1m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cc9ea3d17d74d749065faba20acc346 : RETRY_LIMIT in 2m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab55d9cada094d16a70a01ebe8e21f27 : RETRY_LIMIT in 1m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e57e8f630ea44f98ec0cfdcdd6ebab2 : RETRY_LIMIT in 2m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/be1e6e2b73b84c798fc97125db2dd002 : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/658fd824ffff4e6789c7bdf8735109bf : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/390d0f1ae2db401ca41a1d55ff85f5ed : SUCCESS in 1h 43m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/012ab1529d4f40339439fdf74f925057 : SUCCESS in 13m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/504da9ea8e174b188893361796d6a544 : SUCCESS in 26m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6053f6e8a53442debe64482283eb0a79 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":2},{"id":"60a8313a9e91d2e4bb86afcc6019ab1a010fdc39","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-21 14:34:06.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"c846afc10de8af5ed13491172ec115d0e676f9c2","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-08-21 15:56:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fac2d0706a34df733e5f7efc00a584e6db8ad76d","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-08-22 05:29:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d4460925961ecc4f4b9f0fab96d1d62f12be41b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 07:29:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d0d95f556c74de3bf2b6f5675d740a0 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010727148db249f88974e759600fed36 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6affc59fc434b41bab258a0e4769376 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4dbd4f8165e4f848b8432933888b6e1 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e6d11b16ff4bc99014d24bd2512ab6 : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63fe29c28bef47418b1d7f4775b0fcad : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/718e77484f074b2b85f814691f6e4feb : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65c044e9849b45a2aebe285499f903cf : SUCCESS in 20m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/26bd4f8903624207b221cd91df5488f8 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/af8152f328bd48219dfb088619e1a516 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/70790481497f4ed98ff84b83ce79d615 : SUCCESS in 1h 58m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/01746d3eb6414b3da37eabee609b441b : SUCCESS in 15m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f462d2d9b19547b48a5895e9c151df04 : SUCCESS in 31m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3c54de35278c4a7e97b2d1850b8406c5 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5416f1d920d116ccc5a927a5243b3c41b721c149","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-24 13:07:22.000000000","message":"Patch Set 3: Code-Review-1\n\nShouldn\u0027t we include a release note?","accounts_in_message":[],"_revision_number":3},{"id":"39a78f301e1d8a3e843af8f10bea02028e7ec873","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 13:52:27.000000000","message":"Patch Set 3:\n\nKayobe never did anything support Xen, we\u0027re really just syncing globals.yml","accounts_in_message":[],"_revision_number":3},{"id":"9657238907895801988665246ad873473253e908","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-24 14:04:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3cf306f225feac6d101a2b9e9a77d7b5eddcdcd1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-26 10:27:13.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks Mark","accounts_in_message":[],"_revision_number":3},{"id":"4d402ffd3dc2a39b9c49515d2a5446d7b80b31a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-26 10:42:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ebcbe75aa7f2a972a5d8bb5ff71d72e1c78c8d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:42:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd87741b9080ee6ce779158813f2f65bd9c3a047","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:44:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"83046ec8b2aee408f1d4c1a61d69cb4e1cc178a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:44:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e8917c17a8478f8c5e5fab70475698 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcbf03b9a9d44176aa9560d7c10e5e8a : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6efb266627254b868449a25d72c352c6 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20cddf858934a4db65bd837b6e4e9d9 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6015dc318f9c439ea54ff8ce50314883 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dab740305c21449aae097b2f77940df4 : SUCCESS in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59b7e6fece8d4796ac2bd563536324b6 : SUCCESS in 20m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/098e806f14694589988fcedf4d6ce3d6 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dc34ba7400ad42eea24fc6c99961a481 : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/795680f653dc40ea84d97572ff10cb1f : SUCCESS in 1h 59m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d6ce799e81bc436abc6df246c996ce7e : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/34202824bc41408cb45927b2d13ea93b : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c5f8a9e7c7bd4916848eec96c92dd408 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b53afab4449e4626962998d7dbc0a73ef13f056f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:46:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e848abbf014325accaa845e0db541c : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"597263e8046ac105a86f52314bbb9024b97050c3","revisions":{"4b1c262d883ba41fcdf6e7b3595e6d289d566153":{"kind":"REWORK","_number":1,"created":"2020-08-19 16:00:31.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/71/746971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/746971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/746971/1"}}},"commit":{"parents":[{"commit":"f6f08df6b14b2a575a597f01097c368d33d942dc","subject":"Block Ansible 2.9.12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6f08df6b14b2a575a597f01097c368d33d942dc"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-19 15:36:09.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-19 16:00:07.000000000","tz":480},"subject":"Remove XenAPI integration","message":"Remove XenAPI integration\n\nthe XenAPI is removed from kolla-ansible[1], this PS to remove it from kayobe\n\n[1]: https://github.com/openstack/kolla-ansible/commit/60e03d7bf30f536812fb3f61fb5f1a1db53851f8\n\nChange-Id: I3de408d384141eee1a088d222dca7d583251a298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b1c262d883ba41fcdf6e7b3595e6d289d566153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b1c262d883ba41fcdf6e7b3595e6d289d566153"}]},"branch":"refs/heads/master"},"4df8f06c343a6b222739ce53f3706b2d4ecda3a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-20 14:49:02.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/71/746971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/746971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/746971/2"}}},"commit":{"parents":[{"commit":"a33ba678df211c4d3f53c7ee9b066becd0e962b0","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a33ba678df211c4d3f53c7ee9b066becd0e962b0"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-19 15:36:09.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-20 14:49:02.000000000","tz":0},"subject":"Remove XenAPI integration","message":"Remove XenAPI integration\n\nthe XenAPI is removed from kolla-ansible[1], this PS to remove it from kayobe\n\n[1]: https://github.com/openstack/kolla-ansible/commit/60e03d7bf30f536812fb3f61fb5f1a1db53851f8\n\nChange-Id: I3de408d384141eee1a088d222dca7d583251a298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4df8f06c343a6b222739ce53f3706b2d4ecda3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4df8f06c343a6b222739ce53f3706b2d4ecda3a4"}]},"branch":"refs/heads/master"},"597263e8046ac105a86f52314bbb9024b97050c3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-22 05:29:06.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/71/746971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/746971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/746971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/746971/3"}}},"commit":{"parents":[{"commit":"faf03dbf35386c68202d98920ac92bd2818c6f58","subject":"Merge \"Add support for custom Aodh configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faf03dbf35386c68202d98920ac92bd2818c6f58"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-19 15:36:09.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-08-22 05:29:06.000000000","tz":0},"subject":"Remove XenAPI integration","message":"Remove XenAPI integration\n\nthe XenAPI is removed from kolla-ansible[1], this PS to remove it from kayobe\n\n[1]: https://github.com/openstack/kolla-ansible/commit/60e03d7bf30f536812fb3f61fb5f1a1db53851f8\n\nChange-Id: I3de408d384141eee1a088d222dca7d583251a298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/597263e8046ac105a86f52314bbb9024b97050c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/597263e8046ac105a86f52314bbb9024b97050c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
