)]}'
{"id":"openstack%2Fkayobe~747384","triplet_id":"openstack%2Fkayobe~master~I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e","project":"openstack/kayobe","branch":"master","topic":"ansible-lint-431","hashtags":[],"change_id":"I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e","subject":"Fix ansible-lint after upgrade to 4.3.1","status":"MERGED","created":"2020-08-21 10:27:46.000000000","updated":"2020-08-21 16:34:48.000000000","submitted":"2020-08-21 16:33:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747384-1598027602376-0b925b31","meta_rev_id":"381a802534a6ccc4ab9cd54dd7ed93db4f635b88","_number":747384,"virtual_id_number":747384,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 16:33:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-08-21 13:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-08-21 13:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-21 14:28:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 13:09:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-21 13:10:35.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-08-21 14:28:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-21 16:33:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abc54349654ef8a416f973b7659785df036f4934","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 10:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d7dc1b66775e356a0e30d1b235b437c55e1a082","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 11:08:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f74ea7f14afa59b77d66afda649bc1077c83d16","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 12:40:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cb6004bfa58811ec6f3feae5172aa6d35a9ac67","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 12:51:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2f4a0042764cd4992f8b140e55d582b439177e2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-21 13:09:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8a902d070abfa53f8c84baa0e06cfe2f8eb3bfd8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-08-21 13:10:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ce307805d50e0481975bd10ce6d8002345a1f5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:21:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bc52ece45674fbf9f5f2a031b8482ea : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608c1dbe753c42a186d4b9bfddc2c755 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e6c02cb9de4b49a68aea4b0cb7cb5c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1e96c8385e4948b6fcb9a74d04b896 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d284d9b8d4d94c949024f1cd084d1728 : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6daf6819a30944468b701116bc459de1 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8970869c576c4aa6b42c01e9c9b3eadd : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d631706b4ed547e084331e856f8b48c6 : SUCCESS in 22m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2724597ba2e24f1cbdebdbf28faf3efa : SUCCESS in 1h 23m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/af6607f06fe74a3e8234b248acb88d89 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3e613c8074e949bf839d87603292f084 : SUCCESS in 1h 23m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/45e99b58970545da99622986a6881903 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/24a054db558349c88dfe2140be550520 : SUCCESS in 35m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a7c46ae90a91485da0db692f9cc2a0c4 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7501058d4120f94af6f624eb9c6666b736d5b1e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-21 14:28:25.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"88f474e80c940cd984b2eddcebcbf8b6eaedf69c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:28:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"344fcdd6c24a0e31ca10ff70adad3e54d6ca4ae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:33:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb59922061a94722b4d4c05a08604577 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d67c117a0884b0e94dd9155147b3708 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff12602951a427eb3817940306f4549 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04ec2f6e1cc44638ab2ece450f04de4 : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02541e1c7d27493dbd52a33cc0daec38 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56f4f50affd048a5a72841d777189387 : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e8ccea28da845d28cf8fd9eaa41b133 : SUCCESS in 21m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/755dfccc110f4c9ca184404398e0dcb8 : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb385cfbd52e48c3b53410b58516f100 : SUCCESS in 18m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/85d2f792c7b64bbdbe37ae7378f790ff : SUCCESS in 2h 02m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/915b384e721b4f5db2ccb2bb56f5e7a2 : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/143e44d69417474a8db987ad7ca31ce2 : SUCCESS in 29m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e8fe39f75fb34fd6b3d0024c250968ac : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c32103112c88338357d95293a1dcd8943ae5485d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:33:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"13f68a50e99e60668eee664acfd8687a6971b266","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:34:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a87aee4f6d43198a56074a8ef5f253 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e","revisions":{"67b5300c888959b8f4c6387d3aebfcf845d237d4":{"kind":"REWORK","_number":1,"created":"2020-08-21 10:27:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/747384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/747384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/747384/1"}}},"commit":{"parents":[{"commit":"a33ba678df211c4d3f53c7ee9b066becd0e962b0","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a33ba678df211c4d3f53c7ee9b066becd0e962b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 10:27:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 10:27:10.000000000","tz":120},"subject":"Fix ansible-lint after upgrade to 4.3.1","message":"Fix ansible-lint after upgrade to 4.3.1\n\nChange-Id: I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67b5300c888959b8f4c6387d3aebfcf845d237d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67b5300c888959b8f4c6387d3aebfcf845d237d4"}]},"branch":"refs/heads/master"},"dd6d761f2a807570091f15f616f5542e5149c885":{"kind":"REWORK","_number":2,"created":"2020-08-21 11:08:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/747384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/747384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/747384/2"}}},"commit":{"parents":[{"commit":"a33ba678df211c4d3f53c7ee9b066becd0e962b0","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a33ba678df211c4d3f53c7ee9b066becd0e962b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 11:08:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 11:08:11.000000000","tz":120},"subject":"Fix ansible-lint after upgrade to 4.3.1","message":"Fix ansible-lint after upgrade to 4.3.1\n\nChange-Id: I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd6d761f2a807570091f15f616f5542e5149c885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd6d761f2a807570091f15f616f5542e5149c885"}]},"branch":"refs/heads/master"},"5c78de58104d80ab5ca19852d51a99b2cadf68ff":{"kind":"REWORK","_number":3,"created":"2020-08-21 12:40:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/747384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/747384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/747384/3"}}},"commit":{"parents":[{"commit":"a33ba678df211c4d3f53c7ee9b066becd0e962b0","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a33ba678df211c4d3f53c7ee9b066becd0e962b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 12:39:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 12:39:53.000000000","tz":120},"subject":"Fix ansible-lint after upgrade to 4.3.1","message":"Fix ansible-lint after upgrade to 4.3.1\n\nChange-Id: I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c78de58104d80ab5ca19852d51a99b2cadf68ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c78de58104d80ab5ca19852d51a99b2cadf68ff"}]},"branch":"refs/heads/master"},"f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e":{"kind":"REWORK","_number":4,"created":"2020-08-21 12:51:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/747384/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/747384/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/747384/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/747384/4"}}},"commit":{"parents":[{"commit":"a33ba678df211c4d3f53c7ee9b066becd0e962b0","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a33ba678df211c4d3f53c7ee9b066becd0e962b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 12:48:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 12:48:07.000000000","tz":120},"subject":"Fix ansible-lint after upgrade to 4.3.1","message":"Fix ansible-lint after upgrade to 4.3.1\n\n- Skip rule 106 which fails due to the hyphen in \"kolla-openstack\"\n- Fix code failing \"File permissions not mentioned\" rule\n\nChange-Id: I7ce8a591b4e1eeef749c439e1140ecfb74dcb40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40dde18dfcfc5346c98ed64dfe2dcb58f9dbc2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
