)]}'
{"id":"openstack%2Fkayobe~747653","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~If435e9025f3d78364b229bbe4458f5f1e1ab49aa","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"If435e9025f3d78364b229bbe4458f5f1e1ab49aa","subject":"Make pip_index_url variable conditional","status":"MERGED","created":"2020-08-24 09:07:45.000000000","updated":"2020-08-24 21:47:20.000000000","submitted":"2020-08-24 21:45:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747653-1598305521399-1545f216","meta_rev_id":"44b9980e22f8ab1dd6e1a5b16e4a5c78c002d5e8","_number":747653,"virtual_id_number":747653,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 21:45:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-08-24 19:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-08-24 09:08:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-24 09:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-08-24 19:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 09:08:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-08-24 09:12:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-24 21:45:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b471fc56846cc79efe46f2f5d028c185b8027719","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 09:07:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"788181365c19bbc3e0b89e737fb8a5aa9021c95f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-24 09:08:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b943a1238a976548d407af55bddf492b17ac243c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-24 09:12:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b17ed8f72beca2cc7daab847b112bb5590ae9494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:16:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20d83e29f5d84566a11f9fca1a503638 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a800a12a06734111b18bc05429eabe53 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d468566db0ef48568376f9bbd7c626ea : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9b4a1334934fdcbc235e49d3bad144 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d56be7867114542b7f472683494b452 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7236178b903340c8866706ff9d91b1da : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26c379ef460745c59d12dfcb10aa9601 : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59b766d356434beeae6dd34316e2688b : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87314ed9a3d84a3d8bfbb79570e0626e : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dead543e9c904f2ca8755b744b68d0fc : SUCCESS in 20m 51s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7d1511b2d6b040c98863931708533b33 : SUCCESS in 1h 17m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0c8f1162ccf84d15a3746f9698737cd9 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c337a6c06b0540409eff32d21fc6a581 : SUCCESS in 15m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bf5de374c7604ca39c626bd53c22219a : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/cc057dc8d6d14587bfe549e3ffd622c1 : SUCCESS in 1h 59m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ff2009b1280f43908dc028a7c3c35955 : SUCCESS in 20m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/78e485a741a74f31a5d8fc09886f0a88 : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/04eb584fd8954732be1f1040a394d0e9 : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1abd532cd803c744c5f9005ea99b83ad008db555","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 19:36:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"000d0fe1cf5d4c947ae0f8223c380d06754bca6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 19:36:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edc8bcb92dd338339f3a567f3fe693439882431f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:45:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/193622fe8e984d2e894315442fca3ba8 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a1f842ed09144c59fa1d3cbf43a8a20 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9062cf94b8244334a9f2751179b5b8d6 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0c32d833e354f2780b406e4bdd35b7f : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc7483dbc774947ad42c710a58da21c : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/517523b0979d4eabab219672d2c01903 : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbd8f1b927ee4c0697e10977b4a69234 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bcdf47136ade414e8025b2fe68b49bb4 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/061bb29cc27f4102a2af10a1f53236f6 : SUCCESS in 24m 28s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b7361e368fe24832b988ae522a06f9c7 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d7278d8d295e4e3f9dba6439b6c33e58 : SUCCESS in 1h 27m 57s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/84dddd3131bd4a9990335645e5423efa : SUCCESS in 15m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1c20b8d658434c9e8fdeac014bac8356 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b8a55e1627964d02a66c375a9de10768 : SUCCESS in 1h 58m 06s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a6fcb68b33b74fa5aa6460955e2c9109 : SUCCESS in 16m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/de7118b159a24fb0a435bf89bc50b9e5 : SUCCESS in 19m 33s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b303e8ac523848d588632632cd3d5f72 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1718687cbb23faa5c51edab565a942ccc8b8ea76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:45:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44b9980e22f8ab1dd6e1a5b16e4a5c78c002d5e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:47:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14abfe01034c4c06964e32b76a17ea66 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67673526f10e46a88e02662c7179484c : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e9808a072a4a30d3398f69c560cccc4ff662f6c","revisions":{"9e9808a072a4a30d3398f69c560cccc4ff662f6c":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:07:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/747653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/747653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/747653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/747653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/747653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/747653/1"}}},"commit":{"parents":[{"commit":"28cf998a204a87a03d735c53ef5d7c3b107fb969","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28cf998a204a87a03d735c53ef5d7c3b107fb969"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-08-14 14:18:58.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-24 09:07:26.000000000","tz":60},"subject":"Make pip_index_url variable conditional","message":"Make pip_index_url variable conditional\n\nSetting pip\u0027s proxy variable, operator needs to set also pip_index_url,\neven when one wants to use default value of index_url.\nThis patch sets pip\u0027s index_url only when pip_index_url is not empty.\n\nChange-Id: If435e9025f3d78364b229bbe4458f5f1e1ab49aa\n(cherry picked from commit 7113ae46171ddcbe6d195790b08acbb5c49dfd41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e9808a072a4a30d3398f69c560cccc4ff662f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e9808a072a4a30d3398f69c560cccc4ff662f6c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
