)]}'
{"id":"openstack%2Fkayobe~747654","triplet_id":"openstack%2Fkayobe~stable%2Fstein~If435e9025f3d78364b229bbe4458f5f1e1ab49aa","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"If435e9025f3d78364b229bbe4458f5f1e1ab49aa","subject":"Make pip_index_url variable conditional","status":"MERGED","created":"2020-08-24 09:08:05.000000000","updated":"2020-08-24 21:41:15.000000000","submitted":"2020-08-24 21:39:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747654-1598305169536-038e877c","meta_rev_id":"f58672eaf36e7164f31bb09d7bf35e53d143d745","_number":747654,"virtual_id_number":747654,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 21:39:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-08-24 19:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-08-24 09:08:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-24 09:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-08-24 19:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 09:08:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-08-24 09:12:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-24 21:39:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca2146c688288aabb00ceb686961d4c9acde8c7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 09:08:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"480e4be1e03aecdf65f2d8b0b7915a8671e1e23f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-24 09:08:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"69d6bf484c920d25ae965a2448467213ff72393d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-24 09:12:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6577efdf446351e584ac7db1cea4179fd79b011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 11:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e1a53ba68744a0daf78d990843cfa66 : SUCCESS in 7m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b4722b3ea4341dabaff1c8e6d75df1c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba8c3d80f87a4fb1bc2ef697813d1266 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ac22896049540c4b20e9ea2ca5846a1 : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5338679401041959d3f3205cd7a8883 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e683b892016477aa5fa02158976cf5d : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43fac629d1d445c488923152d3e17fee : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f985baea6f064f85ac336dd4ff767c69 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dcac89095e934ffb8c5f0b298d05be4a : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac0248d7d6d3445eb35aca94b7d49b66 : SUCCESS in 19m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ff83e7c9c8214bb280f476a95b61ae73 : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4de07b31be3a4cf584b2be26c9cd6922 : SUCCESS in 1h 48m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2b37f8005ec44dfcbac0a73f70fc75c2 : SUCCESS in 24m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d2dcdfddc89f44239aa5f9bdf102c040 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"eef62805a181cacba241186a55f16fd1e27ee652","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 19:36:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d9e55b200ca1a27cf6bdab09083a5b3864a969a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 19:36:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a21506d1750a6a3febb1c46ac42538f69d5f42ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:39:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e9aec300b9c54cad860aa58ebb09f91b : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6e15f55aeab4979b4e863309479dd70 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e1705a013c444b7a496f751ca23f32d : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/288934be52b54cce8f6232d5679c666e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2a61126dcb43a4837ea8ef38274cd5 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8f3e5e344d45e7bbd5580ea3766d76 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c61048320e5c4e19a535bd07506ef709 : SUCCESS in 8m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/683b02184c1e49d69b54e39b7f061360 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee7addd24ddd468c908fce8217b80417 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/616521151db74e6095d69948b3d418de : SUCCESS in 46m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2d85ca3c3bb64bd0aafd3395bc3ac6dd : SUCCESS in 1h 55m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/061db87c78b44f22a29c75eec2a84e22 : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/803f901201f742dca3a8d54f0798e7fb : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4b88eb463c30e28b50bb60efbdea67c081f3aae2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:39:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f58672eaf36e7164f31bb09d7bf35e53d143d745","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 21:41:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79144f2de3934a3c85e37c8dfc15c83a : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3250d5ba99aa4cdaafdb75ec5dd065c7 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f95f9a2c734cc45aa7a87613b8c93a651e9be414","revisions":{"f95f9a2c734cc45aa7a87613b8c93a651e9be414":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:08:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/747654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/747654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/747654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/747654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/747654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/747654/1"}}},"commit":{"parents":[{"commit":"dc703008d2b4788ca0af61c85e49971fb0a83992","subject":"Merge \"Use Ansible connection reset support in docker role\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc703008d2b4788ca0af61c85e49971fb0a83992"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-08-14 14:18:58.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-24 09:07:55.000000000","tz":60},"subject":"Make pip_index_url variable conditional","message":"Make pip_index_url variable conditional\n\nSetting pip\u0027s proxy variable, operator needs to set also pip_index_url,\neven when one wants to use default value of index_url.\nThis patch sets pip\u0027s index_url only when pip_index_url is not empty.\n\nChange-Id: If435e9025f3d78364b229bbe4458f5f1e1ab49aa\n(cherry picked from commit 7113ae46171ddcbe6d195790b08acbb5c49dfd41)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95f9a2c734cc45aa7a87613b8c93a651e9be414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95f9a2c734cc45aa7a87613b8c93a651e9be414"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
