)]}'
{"id":"openstack%2Fkayobe~747663","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I22057112ff06cb1aec081dfd05b984f048ce0594","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I22057112ff06cb1aec081dfd05b984f048ce0594","subject":"Pin ansible-lint to 4.2.0","status":"MERGED","created":"2020-08-24 09:34:13.000000000","updated":"2020-08-24 14:14:50.000000000","submitted":"2020-08-24 14:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747663-1598278382088-69a35627","meta_rev_id":"b567f44707acdf6265b877998b1131ef709b3a03","_number":747663,"virtual_id_number":747663,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 14:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-24 09:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-08-24 09:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-24 09:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 09:34:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-24 09:56:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-24 14:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5aebb31646de42507481218876857277d2fbd18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-24 09:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e08efdf0e199583dd78e09b6b9e0cf78ebb87bdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-24 09:34:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25d14afaeb019d311affebe808adf77faa92a286","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-24 09:56:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ae8adf4ddb1b12dd7e631672549b9fb93cd20f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/264f38e450e1416da92326f76043b4bd : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa42d94710a149c282af7fb749a46c64 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78a1cc5ffdcc46f5a956a61eb016ae76 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd71dced51484f1aa3436010e47cbd33 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4e006fa92d440d837284eaf9aab751 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21837e5632d5401eb05d087006ef4bc1 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cdd2ebd2f1a4f82a6edffd7d97ada37 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/207de98d85694a6aaaf01afc64ae4592 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9077378507074bb2a98567887793b7f7 : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1fdae923425d48198407632ddb3f0b4b : SUCCESS in 1h 16m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6061337574e0490991f8f2030a61546b : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/39b0ea8658e743b7b1d10ff1e13c2d38 : SUCCESS in 2h 12m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/959554837cb34b938b0cd7d4c8f2be34 : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/14b562de2e7543c5aa9ef0cf38949598 : SUCCESS in 28m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/13bb02be737d4217975ee0c1a6977afa : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"df0323e1d6e08ac73d448c8bbe8b02d2b01a4238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:01:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"daa67dbb68691dec4f86becdd42cabccd851b5a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:13:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9752ceedec7b4d0398ffdf071f6fe8da : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24f4576b122a4aecaa1c187a12e039f7 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38ca36318dc9469e981e6a8720f79b96 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb866f0c54c459e8d403213a0380a53 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10b5b4108f0343e08a0c212c4383eef5 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b59595769344adfba61d0e123ecc208 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fddb52edc8994a8798b59485d0ddc97d : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a732e867f9a74ed0b7c1d2634d4cab5d : SUCCESS in 23m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/aef3e192f32f47e48088badd3fceafbd : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9ea1e2bb71064049b5bfaffb30e1f03b : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/277ea67a8c664450ae3fed60a71f3307 : SUCCESS in 2h 05m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cd66a8a346904ae9a0bd528e8b03cdc6 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c02d73ff6eaf44a7915ff2aa7ac3150f : SUCCESS in 30m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/05fbce6dab034b888a45fe0157f2071e : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"014c4d1f8052a45afdb4170b4379b60d61e92999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:13:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b567f44707acdf6265b877998b1131ef709b3a03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 14:14:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf44a9e1a0c47abaefb63cdcde57334 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31471bc40bc14c87aa42903e901b3873 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fabc914706abb384717e23ef6b96e10d824c796","revisions":{"9fabc914706abb384717e23ef6b96e10d824c796":{"kind":"REWORK","_number":1,"created":"2020-08-24 09:34:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/747663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/747663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/747663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/747663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/747663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/747663/1"}}},"commit":{"parents":[{"commit":"f82a23089a24df7b6f007e2cc106e12c70143af8","subject":"Skip broken ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f82a23089a24df7b6f007e2cc106e12c70143af8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-24 09:32:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-24 09:32:42.000000000","tz":120},"subject":"Pin ansible-lint to 4.2.0","message":"Pin ansible-lint to 4.2.0\n\nIn stable branches we don\u0027t want to follow the latest rules and have\nconstantly broken CI but rather ensure we control quality checks and act\nas needed.\n\nMost other test requirements are actually pinned via upper-constraints,\nbut ansible-lint is not in global-requirements so it is entirely up to\nus to handle pinning.\n\nChange-Id: I22057112ff06cb1aec081dfd05b984f048ce0594\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fabc914706abb384717e23ef6b96e10d824c796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fabc914706abb384717e23ef6b96e10d824c796"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
