)]}'
{"id":"openstack%2Fkayobe~748161","triplet_id":"openstack%2Fkayobe~master~I93f29ec4334829bfec107ba7aa6df05a47d2ab96","project":"openstack/kayobe","branch":"master","topic":"move_dnf_earlier","hashtags":[],"change_id":"I93f29ec4334829bfec107ba7aa6df05a47d2ab96","subject":"Move dnf.yml earlier in host configure workflow","status":"MERGED","created":"2020-08-26 10:59:06.000000000","updated":"2020-09-14 13:07:02.000000000","submitted":"2020-09-14 13:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748161-1600088738312-b4a3470e","meta_rev_id":"a2f02a9a08c2c23af00013dd26bee2bf66250d7e","_number":748161,"virtual_id_number":748161,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 13:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-09-14 11:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2020-09-14 11:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2020-09-14 11:11:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-11 19:59:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 09:10:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-09-14 11:06:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-14 11:11:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-14 13:05:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46536fdb812249b6f0be77f04bcfaed2f84b1ecb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-26 10:59:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7d61b590d37b415b7c70bacd0e3b577198077d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 12:47:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a169c8382bc5467884264468d68124ed : FAILURE in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c61248aa594da0adddcb5cc76781ef : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb6c076770f43c898be40b0c9a76eee : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/016a89993ce746caa29ea9b6ac9375b0 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8755a7b3b440148868d58191e9f0d3 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64dc5edf11d448b8b15e15acbe7d2d23 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15a921d7e0e5448a8de80d8db8ade783 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b22000b0bcc4c5cb8500d345d0fe62e : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f6217f1478940aabc69bfe8b3faf586 : SUCCESS in 21m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/87195c24e1294bb28d3b82e6eb7d4f61 : SUCCESS in 59m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/52af7055d43b4dcab86b5b56b2f635cf : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4db74da6f8ed4a88b0813c25704fe7f0 : SUCCESS in 1h 44m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2e0160169cfd4126a073a47d221a10d2 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/483d610dd1e546299c4e3fa2d7fb282c : SUCCESS in 24m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4c9dfebdb4cf4b58bba13aacea045e6c : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1ea15d06b704317bfe9c6b9ebe0b0bfba4f89658","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-26 14:23:48.000000000","message":"Patch Set 1:\n\nMakes sense. Does the same apply to the seed and overcloud host configure commands too though?","accounts_in_message":[],"_revision_number":1},{"id":"45f9b61e260e55bf9d056029931be6a9d4acd3d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 07:21:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41dc10c2d4db233e26b4b7362825b938209f01c8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 07:21:43.000000000","message":"Patch Set 2:\n\n\u003e Makes sense. Does the same apply to the seed and overcloud host\n \u003e configure commands too though?\n\nRight, updated.","accounts_in_message":[],"_revision_number":2},{"id":"7143f8d21c90dd023d79fc3da35d335aa92f9296","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 08:44:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ed15159ac9ce99709fb143a056687129de797775","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-28 10:01:31.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d6f07165a84ddbf5d6fbf27dcb96eab9b9fb3eb3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 10:07:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"594f12b7eeaabd1276584967eacc9576bb4421b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:32:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95a9ea1a21f148eea9dce64a892495ce : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5b7d6d625241a392c40c96dbb838e3 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a46ae33904f4f9199d742da3af3c420 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f5b8638b4de4e5fa2278d7975ff9eca : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2d8d61a809424abbb75a3fafa13e1d : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e22c2d8a30444198a1876722af8d453 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1eaab32870a4428e8503497ff8047bec : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ce494b41ccb4d59b30ab9462a7232ac : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d974fc8aa8f44600af0b6e7084d96c13 : SUCCESS in 23m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1924437a9bd0465d97d2d7ebd8eb997a : SUCCESS in 44m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f0919cddf7a74cf8b757d3f8aa49f4b9 : SUCCESS in 12m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1705731e5594947968ed9d2c1dbca68 : SUCCESS in 1h 22m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4370ee6dfdb74dfaa444a35513d8efc3 : SUCCESS in 12m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bd4bd0e34c304a0b993cad7443eddebf : SUCCESS in 22m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1d844b732ae041d686605bd2291f11b9 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":4},{"id":"852f962505f83c6bd44d955a9074cca55a18acc0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-09 09:06:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"321ecc0566ad099362fcfb82a039d10c846e391f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-09 09:07:09.000000000","message":"Patch Set 5: Code-Review+2\n\nI just reordered comments to match the actual order of tasks.","accounts_in_message":[],"_revision_number":5},{"id":"75f4a1e08565a9882ab222065d4dd1d1fdd812f2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-09 09:09:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c304fb40f463c35e35830f7308f5d4a4a2ffb386","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-09 09:10:02.000000000","message":"Patch Set 5: Backport-Candidate+1","accounts_in_message":[],"_revision_number":5},{"id":"c60047dce3f4abd5aed38dea250f5aabda918fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 10:50:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12684526180243c3b4aeb66f18f3c3a9 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f015d90d703142f99497006c68f24179 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98cc161d506247df8586d8176af6c68d : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34964a7899d64ebf9a9ab41230276f19 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f8338477cf413994871a085b5bdcda : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d51b1bb39f5549239bd14e9265b813ad : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e29df48c76e4b05a9b88c746b316d15 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdaf75d493444ecfb99310f207b3bbb3 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7004ae025c1749a9b5abb3cc529e2ac0 : SUCCESS in 18m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/176c27720e564050bbb431ed0dc5db4e : FAILURE in 43m 14s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f0ddca1660ae4abba1629cbe6d721106 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5097445b933b427fa4c3fcda65c6c1cd : SUCCESS in 1h 42m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a1f3df7b7ff944edb06f39c83d6e3656 : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dee7e0cf0554849b14f5e897c0da0e0 : SUCCESS in 21m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6169e11e3f044aa2839cc65845348984 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":5},{"id":"06031c5f831f844b9b95c581cf12a2607f7ecd85","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-09 12:15:06.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0e8050c78c1fda126aff09ef40655d855776795a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:39:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1512b0d2d5d84d5fb58d6ce38f53fcab : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052cb2935ec84d0b886cda8fc4ee0fee : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9efbd488e8246359044ec826b9457f1 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebf22fc1dacf4410a30497a0f0286380 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8978cfa735ea4cd2aa8755fa9e32e191 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a44651f64f1483da03b65d4079d3d3b : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8acc7cf75f34af0a5578b27edc737b8 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a776c17ff884e40880450fbe76d1c91 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57d0e99a4f48491eb9c59593dfed717a : SUCCESS in 21m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b4a051c9a7cf4e1a9cae36b4501c2676 : FAILURE in 52m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/98eef2c2fd1741ea850127b2e120fd97 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/346bc446c11d467d8ce99d3fe0364d60 : SUCCESS in 2h 15m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4a744ad94d464d4a9924b6280440b11e : SUCCESS in 15m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0ecfbbdddc1b424ba6e316d5cff929c0 : SUCCESS in 27m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/53ecf663939a4a28b0425fd8f91c0b67 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":5},{"id":"8246d829f206179f35fbf6b26c06485cd734af83","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-09 19:38:05.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"dd64ffa1efb5e13f26d4fa1270a9c14b7c96f466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:47:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f566b9c2482e4a3aaf5066e1006591af : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4b79a4f3454949aebbd8db5f39da4b : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d1ccc0e471b4433a58cee46ef4dd229 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4551ae99fbcb4f06a1b38b6edd32fd95 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af5f2c378784c4d999f87fa019f4aed : SUCCESS in 16m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99190337566f492aa6f13bd89a2e2669 : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/199e5fb5a1444e6c9984349b845462cf : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b176c2290fa42c285c623a4ec92a557 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c3d2ea9af6d4406b0a40f477de74bc9 : FAILURE in 6m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/61a6c35f73094cde8b65ae8900b7329d : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6ff5169d7bee4f7db618e3d98a0df351 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/65286a9b50c041b39c1fac74aa534834 : SUCCESS in 1h 47m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f2e522617f5042d9b7dcf2bae1db8e10 : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/23779a699b9343549c3ab0a9b1599e18 : SUCCESS in 31m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/de4caa52b03a41338087d4b31017a7a4 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":6},{"id":"3367870393cf687557333123877c32174571b547","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 19:59:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"c7b8068e82499352d27911d749ec886b51916ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 22:03:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d84fc70b630b47adac0c163ad9b5c293 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb311f411374910ade427592cceb539 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93746a48c3b4b8babb67a2e157ca707 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad807e7a0394dd7802406cf838b0dcf : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a6c6cfae8d4d448f47cad502e2b093 : SUCCESS in 18m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dd2b2e0da924e2e8b8870f75c284e73 : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/827b3963188848f1b49e8346d718e4ea : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90089dfa9aaf4de98835281db59b5541 : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5d6f06d158a4d529574111103ce4210 : SUCCESS in 20m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/85a85497ace649a1b7eb9f5b0f53d9ef : SUCCESS in 45m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/17e24b3df2224e5e8b013421d60b859c : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db743b48519e4b6ab7b77301030b029d : SUCCESS in 1h 59m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8d1f6a9c849c4af5925b3c5487b918a2 : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/01a474ea14ff4658b45ca4035aadee05 : SUCCESS in 25m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/eac4590fb1894a6a8c2f180572546cf2 : SUCCESS in 11m 54s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"3d1def9ba3c231cbcb2c5f41693a206d89762cf2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 11:06:09.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":7},{"id":"3b5958adcc10c219965bc4cec8452cfa522560b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-14 11:11:55.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"72da0af86ba347c0007391239b62bf9f7364cdde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:12:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d4324db93a180bf63d0b5f46528df19b804f9810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:05:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca8261811654315b75125b62d6a9f73 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62866d4b14dc4ce4bd748168e45e4285 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d06f2cb845e4c83ae5dd2c57d0c2473 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a118e34c46c484593a0b23580f5655a : SUCCESS in 15m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c6bd4cbe9a547f3aed014017a468346 : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f08d814e83344b88ceb1256e931a832 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/768cdb62d9bf4594b95a1a08c148f6a4 : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4acc479028b4426aae7eb8016a7e809 : SUCCESS in 17m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a581dfb553384711a118f8659f9d7b1a : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0020a3a5871b4de48865c6960ff48240 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc5aa35d0d044feeaec273bd2295407f : SUCCESS in 1h 46m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/881a4cd4d59a4d91b62860a1a2496358 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8f90904fa4584c0b8070922effffd9d8 : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6f471906cd394b859fab21a071fcf9ed : SUCCESS in 11m 54s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"045b46a5b54ff1bb6eec858e1d4b57392b9c4791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:05:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5cba15a21ba2e42a30bd04fe88e0a6efb7328934","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 13:07:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0176c546dd8a4b8b8181cef393314613 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51a310f4d06479a8611b867ef47a2c2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"10943580334cc654ecb61cc446ef7c1f61abe38c","revisions":{"49f525ed07d8bef699750b2aeca3c91c9babdd17":{"kind":"REWORK","_number":1,"created":"2020-08-26 10:59:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/748161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/1"}}},"commit":{"parents":[{"commit":"49885a2b8dcd7b9bd214cf597030caf76e2d5334","subject":"Install bash-completion in dev-tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49885a2b8dcd7b9bd214cf597030caf76e2d5334"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:59:03.000000000","tz":120},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set later,\nit fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f525ed07d8bef699750b2aeca3c91c9babdd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f525ed07d8bef699750b2aeca3c91c9babdd17"}]},"branch":"refs/heads/master"},"70f2550883a91a34528694d00191914f4096f9d0":{"kind":"REWORK","_number":2,"created":"2020-08-28 07:21:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/748161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/2"}}},"commit":{"parents":[{"commit":"49885a2b8dcd7b9bd214cf597030caf76e2d5334","subject":"Install bash-completion in dev-tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49885a2b8dcd7b9bd214cf597030caf76e2d5334"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-28 07:21:03.000000000","tz":120},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set later,\nit fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70f2550883a91a34528694d00191914f4096f9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70f2550883a91a34528694d00191914f4096f9d0"}]},"branch":"refs/heads/master"},"2e175df88768678d84f9cacfd624065f9e69a025":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-28 08:44:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/748161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/3"}}},"commit":{"parents":[{"commit":"49885a2b8dcd7b9bd214cf597030caf76e2d5334","subject":"Install bash-completion in dev-tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49885a2b8dcd7b9bd214cf597030caf76e2d5334"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-28 08:43:56.000000000","tz":0},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e175df88768678d84f9cacfd624065f9e69a025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e175df88768678d84f9cacfd624065f9e69a025"}]},"branch":"refs/heads/master"},"378517e5359a20acde0fcbbc172e4fdbba3e42fe":{"kind":"REWORK","_number":4,"created":"2020-08-28 10:07:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/61/748161/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/4"}}},"commit":{"parents":[{"commit":"49885a2b8dcd7b9bd214cf597030caf76e2d5334","subject":"Install bash-completion in dev-tools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49885a2b8dcd7b9bd214cf597030caf76e2d5334"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-28 10:07:56.000000000","tz":0},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/378517e5359a20acde0fcbbc172e4fdbba3e42fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/378517e5359a20acde0fcbbc172e4fdbba3e42fe"}]},"branch":"refs/heads/master"},"52e83be06269995cccd0cb7dfb1ce2bbc012d832":{"kind":"REWORK","_number":5,"created":"2020-09-09 09:06:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/748161/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/5"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-09 09:05:48.000000000","tz":120},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52e83be06269995cccd0cb7dfb1ce2bbc012d832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52e83be06269995cccd0cb7dfb1ce2bbc012d832"}]},"branch":"refs/heads/master"},"a0db664f3c6f458d2ddd98c006c27cb12d551bed":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-09-09 19:38:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/748161/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/6"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-09 19:37:47.000000000","tz":120},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\nDepends-On: https://review.opendev.org/750804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0db664f3c6f458d2ddd98c006c27cb12d551bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0db664f3c6f458d2ddd98c006c27cb12d551bed"}]},"branch":"refs/heads/master"},"10943580334cc654ecb61cc446ef7c1f61abe38c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-11 19:59:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/748161/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/748161/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/748161/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/748161/7"}}},"commit":{"parents":[{"commit":"7b605851349a8c469d11d0f9daacc0f56474d23f","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b605851349a8c469d11d0f9daacc0f56474d23f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 19:59:02.000000000","tz":0},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\nDepends-On: https://review.opendev.org/750804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10943580334cc654ecb61cc446ef7c1f61abe38c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10943580334cc654ecb61cc446ef7c1f61abe38c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
