)]}'
{"id":"openstack%2Fkayobe~748523","triplet_id":"openstack%2Fkayobe~master~Id636f78c61237fb27fa65fa3d4b3fc1a4cf0ba6a","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id636f78c61237fb27fa65fa3d4b3fc1a4cf0ba6a","subject":"docs: Move configuration reference to subpath","status":"MERGED","created":"2020-08-27 19:40:29.000000000","updated":"2020-09-01 10:12:39.000000000","submitted":"2020-09-01 10:09:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748523-1598954966740-69a35627","meta_rev_id":"30b88e96cdb10567118cc74bc86b644f013241bc","_number":748523,"virtual_id_number":748523,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 10:09:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-28 08:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-09-01 07:48:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-28 09:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-09-01 09:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2020-09-01 09:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 08:35:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-28 09:57:52.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-01 07:48:19.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-09-01 09:45:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-01 10:09:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02b2765a8ece015711d5b697daa762f53fc5e9a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-27 19:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29f5e3ad93dc022402c89c30eef4115840719c73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-27 19:40:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a387ff42c9c551664646c1fdb72c92d14866414f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 20:08:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776acf2314cd4a0fbc71a2ff0a9238fa : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a670d681b44854abaa74623889be9e : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10bb4ab15bfe4ed3a0e3fd73f8c54ff0 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c844af5884443e383164ae7c8c52a46 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648b6b329a4b425b8e73ae8c48232e1b : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32aea8a834bf4eb5a4844d1399741390 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f83a6375dbd437abc03b255af72cae3 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f50a5b348857424c9246860a6e5961e6 : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a674549769c14168a449826ad2751868 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a68e1a1ad7067635a55615fd6045a832c95a502b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 06:38:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f6c66a8eb7f04d9f3ea4af37be1990e293533cec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-28 07:42:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f320647a716a2822a5080fd4f0651e633b573b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 08:21:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48cb79a1df8b48c9a157e97ab8590412 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627b8768378c42b0ba50e7f6afcd4402 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d684324ec3142be98a3bfdaf7c0cb27 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba5958d5922f4419b05af53744017613 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53082d23af44b0fbfbb3ce24f9c03e4 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c93e6d2390e4ec2a9b49d750bfe84a3 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2680d5b29e5244a8acd7a6d1c60c48d6 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a77a8ce74f9d48f28e9540e99215e463 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ce98ee740ef42f697e49a069ef4ac79 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":3},{"id":"71e544522df4793a22ac5cadb6727df059145f2d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-28 08:35:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c9d7a5231ced909b5790ca4ae924e20188d06e54","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-28 09:57:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"973c438bb2895f77d5d04c9525610777f4a94168","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-01 07:48:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c0a81442362803d2c4182cd5555c4fbdaca41ad1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-01 09:45:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a27aa39f8248d214eed896940e855adf83bfb4e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:45:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5840f8e5390b24a91766a7f5adb8d5561c613da0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 10:09:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84add154ecc4ab3a537536ee813f736 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f950df644bc3494990de8369558e5ed0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b275575634a7439fb10d109c1cc8acd7 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217dcf660bf44db786a628b6b4be8591 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e59cd3568a3445eaf793802febd5c79 : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79cd7f8c590a42648d45fb058c5a8145 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f367acfe29c048af8b247bb4c7cbc3d5 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfe7b7a80d7c487c913ab8ac6280ca7b : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":3},{"id":"7c4bab9c285fa0eacf47e4f8829f72c0cc9371b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 10:09:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"30b88e96cdb10567118cc74bc86b644f013241bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 10:12:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601b77819b344e658fd7d345e0fd2c24 : SUCCESS in 2m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/317dc46e0eda4fb0a9cd70ebef6d6b66 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"354113c20b419f478d4d6919ebde69aa8d2da6af","revisions":{"e1d02fa5b823f170fc8905dd404210ca706bd449":{"kind":"REWORK","_number":1,"created":"2020-08-27 19:40:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/748523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/748523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/748523/1"}}},"commit":{"parents":[{"commit":"731957c58a48818390888bc65d20081cd53960d8","subject":"docs: Move getting started to a separate page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/731957c58a48818390888bc65d20081cd53960d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-27 19:18:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-27 19:38:20.000000000","tz":60},"subject":"docs: Move configuration reference to subpath","message":"docs: Move configuration reference to subpath\n\nThis allows us to add configuration scenarios.\n\nChange-Id: Id636f78c61237fb27fa65fa3d4b3fc1a4cf0ba6a\nStory: 2004360\nTask: 40777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1d02fa5b823f170fc8905dd404210ca706bd449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1d02fa5b823f170fc8905dd404210ca706bd449"}]},"branch":"refs/heads/master"},"3855a69c134eac8d517200674a8d0f5af0d061c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-27 19:40:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/748523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/748523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/748523/2"}}},"commit":{"parents":[{"commit":"faf03dbf35386c68202d98920ac92bd2818c6f58","subject":"Merge \"Add support for custom Aodh configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faf03dbf35386c68202d98920ac92bd2818c6f58"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-27 19:18:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-27 19:40:38.000000000","tz":60},"subject":"docs: Move configuration reference to subpath","message":"docs: Move configuration reference to subpath\n\nThis allows us to add configuration scenarios.\n\nChange-Id: Id636f78c61237fb27fa65fa3d4b3fc1a4cf0ba6a\nStory: 2004360\nTask: 40777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3855a69c134eac8d517200674a8d0f5af0d061c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3855a69c134eac8d517200674a8d0f5af0d061c9"}]},"branch":"refs/heads/master"},"354113c20b419f478d4d6919ebde69aa8d2da6af":{"kind":"REWORK","_number":3,"created":"2020-08-28 07:42:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/748523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/748523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/748523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/748523/3"}}},"commit":{"parents":[{"commit":"faf03dbf35386c68202d98920ac92bd2818c6f58","subject":"Merge \"Add support for custom Aodh configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faf03dbf35386c68202d98920ac92bd2818c6f58"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-27 19:18:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-28 07:28:40.000000000","tz":60},"subject":"docs: Move configuration reference to subpath","message":"docs: Move configuration reference to subpath\n\nThis allows us to add configuration scenarios.\n\nChange-Id: Id636f78c61237fb27fa65fa3d4b3fc1a4cf0ba6a\nStory: 2004360\nTask: 40777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/354113c20b419f478d4d6919ebde69aa8d2da6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/354113c20b419f478d4d6919ebde69aa8d2da6af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
