)]}'
{"id":"openstack%2Fkayobe~750897","triplet_id":"openstack%2Fkayobe~master~I3442174b378cc6e361550fcb9b32a62f61fa7972","project":"openstack/kayobe","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I3442174b378cc6e361550fcb9b32a62f61fa7972","subject":"Update hacking for Python3","status":"MERGED","created":"2020-09-10 07:53:01.000000000","updated":"2020-09-11 18:17:08.000000000","submitted":"2020-09-11 18:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750897-1599848109277-a980bdbc","meta_rev_id":"fb1e4c4f2c210ea2b23e67c66ae7ba549503bf24","_number":750897,"virtual_id_number":750897,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 18:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-09-11 16:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-09-11 16:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 13:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 12:25:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-11 16:22:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-11 18:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2ca9c17a8a33ba53e9c39f448b467d99745fb30","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-10 07:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"857de9472ff538719c1f3b7633507966ac7c1318","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-10 08:36:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3523ad9bbf0825ee977ff21626ffc213e497acd0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-10 08:48:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac51de4c5f59b3495fb7b55610f8099b7361919a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-10 08:48:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93a6ac4bb9452f62d17207ae847c3318c5c3faaa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-10 09:27:33.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"d5f6f66e6514cc928a5766a742baf09c8a5c4ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:01:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a1276885368415688916e85abb30985 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e60f38af524fd0b4b8c76f0e8abeb6 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d607fd162524fcdb0080f8f89e38577 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8005bb8221a44a2e82754d35d7a3c985 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6707ae9ff8574fb5a715b852cbc6544a : SUCCESS in 16m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a7a3daf12d84c25b49917848299082a : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffe9467afded4ca89973ade4da748d4f : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79718a5cb94347f29f9db5d4e3cadcd5 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c8e601c91c04cdfa5743ed9864aaf88 : SUCCESS in 20m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d6f4cb230d5d47d69f24d907e1de54ac : FAILURE in 51m 32s (non-voting)\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/df1f546829744a4c865fc92122840d6a : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e3dbc41cd5744a1d9e81885da5f1fd64 : SUCCESS in 2h 05m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f2186b7f8a2641bfac2126bdd60e862a : SUCCESS in 17m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3342a9b5e726475f9062ea77f65c7b3e : SUCCESS in 28m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/927a4d908afd4a89834bafb14b11a98b : SUCCESS in 12m 06s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"78a8517422a1e869bdd105fdf62feff3667c0c51","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 09:30:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5afd6b40e9b6c7a198ebcd2e46404f864c5adc3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 10:50:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d59dd53d3f96d4e4e1375e3f3dcf15248a736a7e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 10:50:28.000000000","message":"Patch Set 3: -Code-Review Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"ded5a2b183b4a26a6cd2192efb8e661c1d0163d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 12:14:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d852c3c3ead49749b891f5b23dfa3c2 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5da337233be4389bbe8b1f0fd16a595 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81aa9bf4146348a2969bef0b54bb0377 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09e5ab8eee0d4f229a04967dd6520419 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8fa84f337a442908863133ce2c0c883 : SUCCESS in 14m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c039f2e964204013b3a59f82f177939b : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/545fb7e14dc44f128abcce5643e17feb : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20c15a98ebad47b69c3fa4e6f0e4f53f : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/812fff75d5b8434da78f747c77308603 : SUCCESS in 19m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3d7160ee3f044a4faf5ebd83a3e2da26 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/14e2946bdc1e4887bb38d1e5f0625ce6 : SUCCESS in 18m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2a4c391db109443ea53702c97ae82f79 : TIMED_OUT in 2h 33m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ad8fb52d0b1b41f6be3762536f907644 : SUCCESS in 13m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1a3761a9c8494ccbb5b4af73055c739d : SUCCESS in 20m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6401c1af68a64f539cab69a6c18bd7ec : SUCCESS in 13m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"2fd95e67c2c6bba081ae4ebd97cf7ec3a941f1d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-11 12:25:47.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"287c2bb75f910267b4d9367013c4c267df255a2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 12:37:42.000000000","message":"Patch Set 3:\n\nThere weren\u0027t any issues, the run was just slow, it was executing the following task when it reached the job timeout of 9000 seconds:\n\nTASK [Perform testing of the baremetal machines in the upgraded overcloud]","accounts_in_message":[],"_revision_number":3},{"id":"5b090e1e1da89fad5da3dead4f0eda5d46b022df","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-11 12:59:39.000000000","message":"Patch Set 3:\n\nyeah, these happen with CI from time to time, no need to worry; but now pep8 is failing","accounts_in_message":[],"_revision_number":3},{"id":"5e6ff11b36361d8c25f99b9e9ff97356077b2dc4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-11 12:59:50.000000000","message":"Patch Set 3:\n\nsee https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_6db/750897/3/check/openstack-tox-pep8/6db8dc4/","accounts_in_message":[],"_revision_number":3},{"id":"98531bef43077e98559563e5e455a3a14b46297e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-11 13:00:04.000000000","message":"Patch Set 3:\n\n2020-09-11 12:34:20.409869 | ubuntu-focal |     src/pcremodule.c:32:10: fatal error: pcre.h: No such file or directory\n2020-09-11 12:34:20.409877 | ubuntu-focal |        32 | #include \u003cpcre.h\u003e\n2020-09-11 12:34:20.409884 | ubuntu-focal |           |          ^~~~~~~~\n2020-09-11 12:34:20.409891 | ubuntu-focal |     compilation terminated.\n2020-09-11 12:34:20.409898 | ubuntu-focal |     error: command \u0027x86_64-linux-gnu-gcc\u0027 failed with exit status 1","accounts_in_message":[],"_revision_number":3},{"id":"aba36440655394ea0226bbbb209539159467d1a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 13:20:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e36b530c7b18273279291906395a6d457b8c34f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 13:26:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"15947ab012d8c1ac0da75755c21789cee02e0246","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 13:44:29.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"fd9f25bb5e0064659c028f86e193e1dcf0bbfbd9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-11 14:08:27.000000000","message":"Patch Set 6:\n\nPatch set 5 randomly failed in the kayobe-seed-vm-centos8 job so I submitted another patch to avoid waiting two more hours.","accounts_in_message":[],"_revision_number":6},{"id":"d7879e7bf3f6a4f4c7544d3fcab0df162a48df45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:09:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd167ed4635d49dd86c8d37c44e163db : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ca7649b81a4cd89e18e34c44ba5749 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe603623d0fe43e0a23fbcfe8b5b6f14 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74e8e994e6af4de88b9ad6c5b02847be : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feff3aceb1e6433b8cb3a1be65801ed8 : SUCCESS in 15m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c76430d4eeec4ced882cd9a85c1e5b22 : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/171233f007704b37a2796d796e36ef29 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11c4427701804e37abe8005a68b08e40 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a3dff50d07d421e94ee244a9a32fab4 : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/14811e9644df45ada616eb81dcf4971f : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/96b08a8962254dfe96a46996653e867b : SUCCESS in 18m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f8db7ae5a884f88893718781e18f955 : SUCCESS in 2h 18m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc913948ed704a36ad7c7ef8cb3766a6 : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1c2095da3df1420e9c183f7f77e5f838 : SUCCESS in 25m 41s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/576a88bce7594a9daa91036a7a0d0583 : SUCCESS in 11m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"40288481c7f1c3ec0540d16251d28a7d8ecdf970","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-11 16:22:46.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2a484d29d71d3148361afab691fba5a1cb41ae90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:23:03.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5861fcbe79e39f9e19de2785635c3bd02322e954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:15:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e7c8b9fedb44178252a88dbc7b046c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c295a1580d54d3699aaaf1d3f33fcb9 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d96268287d34453cb9110b6b9286f5dd : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb5d94d648c4c52b63d7db4feb1922e : SUCCESS in 15m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39933bd8d40f4e73877321aca2e33ea9 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b3fe98f2ec544e0a9911bba00484e14 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f60383ea815845479d258e8189fa7382 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9449e913c6dd4c6391551d4405dd7fbe : SUCCESS in 20m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1e1f631a5188490f92e45b7aca1228e2 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b1a998a97394436ab01bea1af885c6e8 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bb1ec6d42ffd48b3b43a01ad382d2ddd : SUCCESS in 1h 47m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c93f0aeb70f34e279e290a4ef173271d : SUCCESS in 13m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/24f935cde3c14674a50bb28452a5d9c0 : SUCCESS in 26m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/41ab6955abb14df788a9f3a5df6dd55a : SUCCESS in 10m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"ddcc1e6bef0ff205d466aab55b9f362f37ce3b7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:15:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2d442748974406e87cd281a292018c1568364842","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:17:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fd96c237944956b8dc5a5ed89abe81 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/194769429d7644e49c729005c2fcbf69 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7b605851349a8c469d11d0f9daacc0f56474d23f","revisions":{"b564c2642134c264a6aae59feb73e0b4f7d4601e":{"kind":"REWORK","_number":1,"created":"2020-09-10 07:53:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/1"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-10 07:51:01.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-10 07:52:48.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nTemporarily disable kayobe-overcloud-centos8 job which is broken by\nkolla-ansible until https://review.opendev.org/#/c/750804/ is merged.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b564c2642134c264a6aae59feb73e0b4f7d4601e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b564c2642134c264a6aae59feb73e0b4f7d4601e"}]},"branch":"refs/heads/master"},"3a211f44241f831b0a51aa578be6886a8ce6aa74":{"kind":"REWORK","_number":2,"created":"2020-09-10 08:48:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/2"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-10 08:48:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-10 08:48:25.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nTemporarily disable kayobe-overcloud-centos8 job which is broken by\nkolla-ansible until https://review.opendev.org/#/c/750804/ is merged.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a211f44241f831b0a51aa578be6886a8ce6aa74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a211f44241f831b0a51aa578be6886a8ce6aa74"}]},"branch":"refs/heads/master"},"b27d2ebb5550892620af7c52395024b55c233870":{"kind":"REWORK","_number":3,"created":"2020-09-11 09:30:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/3"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 09:29:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 09:29:51.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nIgnore bashate failure on one-liner loops.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b27d2ebb5550892620af7c52395024b55c233870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b27d2ebb5550892620af7c52395024b55c233870"}]},"branch":"refs/heads/master"},"ce1a453f9cc232302661867684b5ce88f2efede2":{"kind":"NO_CHANGE","_number":4,"created":"2020-09-11 13:20:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/4"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:20:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:20:24.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nIgnore bashate failure on one-liner loops.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce1a453f9cc232302661867684b5ce88f2efede2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce1a453f9cc232302661867684b5ce88f2efede2"}]},"branch":"refs/heads/master"},"d257e3a51711a008a398e154bca299faafcbadf9":{"kind":"REWORK","_number":5,"created":"2020-09-11 13:26:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/5"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:25:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:25:40.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nIgnore bashate failure on one-liner loops.\n\nAdd pcre devel packages to fix failing pep8 jobs.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d257e3a51711a008a398e154bca299faafcbadf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d257e3a51711a008a398e154bca299faafcbadf9"}]},"branch":"refs/heads/master"},"7b605851349a8c469d11d0f9daacc0f56474d23f":{"kind":"NO_CHANGE","_number":6,"created":"2020-09-11 13:44:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/97/750897/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/750897/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/750897/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/750897/6"}}},"commit":{"parents":[{"commit":"d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06","subject":"Merge \"docs: Move configuration reference to subpath\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3f01bdfd14e8bce2ef008c09ea91e8d6cd04c06"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:44:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-11 13:44:17.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nIgnore bashate failure on one-liner loops.\n\nAdd pcre devel packages to fix failing pep8 jobs.\n\nChange-Id: I3442174b378cc6e361550fcb9b32a62f61fa7972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b605851349a8c469d11d0f9daacc0f56474d23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b605851349a8c469d11d0f9daacc0f56474d23f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
