)]}'
{"id":"openstack%2Fkayobe~751785","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I93f29ec4334829bfec107ba7aa6df05a47d2ab96","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I93f29ec4334829bfec107ba7aa6df05a47d2ab96","subject":"Move dnf.yml earlier in host configure workflow","status":"MERGED","created":"2020-09-14 13:18:13.000000000","updated":"2020-09-14 21:29:14.000000000","submitted":"2020-09-14 21:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751785-1600118860893-309c905f","meta_rev_id":"d67d1540987ba87b90f4dbf2b2402f722b2494d9","_number":751785,"virtual_id_number":751785,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 21:27:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-09-14 18:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-14 19:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-14 19:29:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 13:18:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-09-14 19:29:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-14 21:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cab351f9e2dd1f9fc3d549d47c0c2d39997dc416","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 13:18:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ae241cf7daacaaa66965d98fb150d6137030fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:19:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a0a53fd958c4f2f82ef1816b53bfd56 : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f7dff38109447a5a50fad858c136268 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20aefbff0c1743a69c4be3b9fba125d3 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196806cd4a7c44039d4dd2cfc82b84f5 : SUCCESS in 8m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f22cc7d2a8964cb78c44a556446a726b : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7a2603cd1c45d3a43847b78d9461f6 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf91e1610c4345eea57caaebfb026041 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af236e09ff884e53ba7c386e9510bb22 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/611cec01c8cf4a9fa6df53839b8d84b8 : SUCCESS in 6m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2356662637f34469807cd69517299817 : SUCCESS in 19m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/53d658da59084aaeaefb7681740d1f3d : FAILURE in 5m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3a575b1a6ceb44a7aa3d7328f5873969 : FAILURE in 8m 46s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/1441ae6968e944bdb32883e5ba1366e6 : FAILURE in 11m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4cfe10bb36454fbb872924d4cf418699 : FAILURE in 5m 56s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7ecf2538e6884810ab131efbff6e26f0 : SUCCESS in 1h 57m 26s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/4140b01c7c0c4bb496c2e0c9965b01f7 : FAILURE in 6m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/729012b59c044b86919f5d4a539d1b89 : FAILURE in 6m 15s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/88ac5906c59041d3931eb19f8395b9f7 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7c06760a71d8e7efdd9d29759bafc67e7cdd2485","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 15:52:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b3229779911f92aab937799f98cfc76368613bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 17:56:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8da5e22e53ae4f3d9cd418f17468b6b5 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff1de5a863a24fa19e1bcea91c48842c : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/026220b7bee54688a69a3cdd578f7e25 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfc7fbcac8a48d488b8f9612d14b545 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66e0e6488a334da6be7912adc47cf6d4 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1bf9fe5ba047a79472c76293077338 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf3d7cbc7be14906a4776b1b05cb553a : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5751b2386ba4e02bba46bca387a4cd0 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22b4d298966046fcb589475fb65e5cde : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d559290a716149509cec90ca174dec33 : SUCCESS in 15m 17s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a6a39b9c12584be5968f5fbb71dc16fa : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/02d3e189f92c4b95a893510c7dce7bcb : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5aa7a43012384e369e4217f0d9a27e42 : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/129538bd5ade4a529c31821658d5973d : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ac2129c583364af2808af5382e6cdbea : SUCCESS in 1h 48m 48s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/36b1841ef588446f9df5d623865d5087 : SUCCESS in 29m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/68126d22532142f9883d8a0ed69f1bbf : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/98b407c98402486e8e4ff85e3bc86ab2 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3d349852ae9e7fbace36810b9b0b3961b8307b37","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 18:05:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"32fa415e259980af3393c75ed26fc67250b9233a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-14 19:29:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e5a5f491db6320326dca48159bff90898526bbc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 19:29:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d0a2b8d0d6bff375ea90a4f6e1ec1144539b9c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 21:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c3be5377b830743048bc8554b5140fb274056c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 21:27:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47f399bf412e461cbfcc82e64d1940db : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d21b0077bdb044d8a8d77094b2557003 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b300f36a674c9280973cd6362ee104 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3489a98a106b4900ad90a4e5c8b263f0 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44fd4d2df0840109e64cbab550eaf54 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7cf3602a03a4aca8e20a5628ffa4e6f : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4a580a6d5724b30b953d662338638f0 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e82be51ed55a45b595ce175f01d27ca3 : SUCCESS in 7m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/478e161d699c4590ad353b9422cba595 : SUCCESS in 19m 10s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9ba4eb973a3742c4ab726f6b9c34b757 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c1601eea65764e26b5f631bcc268cb9c : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/669de6d6ba8f452cb6cd177486d2bcbb : SUCCESS in 14m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2ae4b71de6b245b999ca42da3e59d09d : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8dfeb97c0b3544dbb605b0714f9bea93 : SUCCESS in 1h 53m 29s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/02bbacedb7944944a8faa2a62550e4ee : SUCCESS in 20m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab0a8b83bd3748fabbcf380a71ecbd0a : SUCCESS in 18m 25s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/567fceaa715e4402a960f24b380f23cd : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d67d1540987ba87b90f4dbf2b2402f722b2494d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 21:29:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71d094ba72d428fa2f5392616dbbfb1 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e5463c63b01462191b932627f54e136 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac","revisions":{"6dbaa5181185c8604d50e3b525bddc65248e2004":{"kind":"REWORK","_number":1,"created":"2020-09-14 13:18:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/751785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/751785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/751785/1"}}},"commit":{"parents":[{"commit":"36e27c6d317a8fc32f1968914546408eec5c2bbf","subject":"Merge \"Set endpoint type variables to publicURL in public-openrc.sh\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36e27c6d317a8fc32f1968914546408eec5c2bbf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 13:17:47.000000000","tz":60},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\nDepends-On: https://review.opendev.org/750804\n(cherry picked from commit 10943580334cc654ecb61cc446ef7c1f61abe38c)\n(cherry picked from commit 53420e06d4ef78974474e138e8fbfb9a052d60ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dbaa5181185c8604d50e3b525bddc65248e2004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dbaa5181185c8604d50e3b525bddc65248e2004"}]},"branch":"refs/heads/stable/train"},"739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac":{"kind":"REWORK","_number":2,"created":"2020-09-14 15:52:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/751785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/751785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/751785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/751785/2"}}},"commit":{"parents":[{"commit":"36e27c6d317a8fc32f1968914546408eec5c2bbf","subject":"Merge \"Set endpoint type variables to publicURL in public-openrc.sh\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36e27c6d317a8fc32f1968914546408eec5c2bbf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-08-26 10:46:08.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-14 15:52:50.000000000","tz":0},"subject":"Move dnf.yml earlier in host configure workflow","message":"Move dnf.yml earlier in host configure workflow\n\nSince kayobe is installing python3-pip in pip.yml - and dnf proxy is set\nlater, it fails on no-direct-internet systems.\n\nChange-Id: I93f29ec4334829bfec107ba7aa6df05a47d2ab96\nStory: 2008067\nTask: 40750\nDepends-On: https://review.opendev.org/748587\nDepends-On: https://review.opendev.org/750804\n(cherry picked from commit 10943580334cc654ecb61cc446ef7c1f61abe38c)\n(cherry picked from commit 53420e06d4ef78974474e138e8fbfb9a052d60ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
