)]}'
{"id":"openstack%2Fkayobe~752472","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I908dc3320cb1dce7a040a1dbfeac414ace22d6e3","project":"openstack/kayobe","branch":"stable/train","topic":"story/2007900-stable/train","hashtags":[],"change_id":"I908dc3320cb1dce7a040a1dbfeac414ace22d6e3","subject":"Avoid setting IPADDR\u003d0.0.0.0","status":"MERGED","created":"2020-09-17 13:51:50.000000000","updated":"2020-09-18 11:04:41.000000000","submitted":"2020-09-18 11:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752472-1600426955168-309c905f","meta_rev_id":"3f75c15cce3744d5182f3b937d2a7a6c498717e9","_number":752472,"virtual_id_number":752472,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 11:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-18 07:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 08:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 08:45:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 08:45:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-18 11:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d337275185eaaa19122a51da372a22dcb1a4c32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-17 13:51:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0809d637ab73797f7b1979c54b073abc210a01fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 16:07:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f7f8c1be7aa40aca10a19663d48c30a : SUCCESS in 7m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e14a66f7086245d6b97478fb852e2bd7 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2f686654b7b44b09f77338f14df6e1e : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24185d34cae94b59b53cb824c5f198ad : SUCCESS in 13m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c64601b2d7d149a7ae626bc19d41fe28 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8ebfa3b88549a68ca2f5ab3b9c3a67 : POST_FAILURE in 42m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33eadac1090242d5978b0d4765905426 : SUCCESS in 8m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74261513413341fda40f5d3f4677ed21 : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbf952f9c2ed492e99192d8c2f83da89 : SUCCESS in 9m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0350336eee7346c8a343ebc67b21ddaa : SUCCESS in 19m 57s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6e53a11dc8fa4ac0aca2f6e05afbfa45 : SUCCESS in 1h 10m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7221528754c94716bb55ad7906e35e83 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/622d9bac51a84a6787006b462cce76d3 : SUCCESS in 15m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f18e0133351243ecb9bac90bd2e0e31b : RETRY_LIMIT in 11m 38s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2f03d6263c564262bb3bdf5f7b24d997 : SUCCESS in 2h 00m 06s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/496139d3ceca4ea1b96e0874dc900586 : SUCCESS in 25m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/920c9744fd2547c59f9bbf526ca800f9 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f8e9c2cd5b6a46c7a7e54b586b3f8535 : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"73e38f620abfd1784d02cbe26441749665ed7f79","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-17 16:43:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5cfbcf421aa2a59e4f9ff677b4633fe276b93401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 18:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a231cb3eb9c447a967770fc4c34cf1f : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5252ce9db95541639f9500cf22bbd0d0 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef2724fee6ba4cdba9ac086fc27cd74d : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056caae45f1d42c981c27c3efe913469 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/263852d52970428a8221b7943e10e43c : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eaacefcf7a84de0ad6c45f0ffd2f208 : SUCCESS in 20m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ea20f103b0d45559885760f68f70817 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea5143c5d24a4219981d036ddbe59b99 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee26d7e9ab504c9f95c4efb504ffd043 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e023e92213f4eb5b2f0f767f6181852 : SUCCESS in 18m 26s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/84f149a1221243bf971cd879c9006b2f : SUCCESS in 1h 17m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ee637a15dca4076818a56bd3ba09158 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c207ae2047494498b4c5ac0dedf1fa11 : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7561bd9e07b04533bd8a2202c13775c1 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/94ffd43a2aae4edfb77caaf356d068c8 : SUCCESS in 2h 03m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/036791ecf0b34a6c9edb75438f00e1b2 : SUCCESS in 21m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ecdc900cd8454a8eb51daec73443fd55 : SUCCESS in 22m 30s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/61e9574b18a943b68726bd91db367c2f : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"800aa561a9cdc209bb944c4910edb567c04fba80","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-18 07:36:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2eb9b851a1a3763cc5bacef7078f7cfda28b266e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-18 08:45:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1f3921ca0b892aba54e13bf531a198f2acb2c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30e1efb4cd308186c2e95efe6d517e46e4c3bf3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:02:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/188994475eec4fd09eb52ccb294fd8da : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca2f3cc5aef447a8baa3459c12d334eb : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c63e2d8ac0074c958271cfae7aa4c10e : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/719bb9266a5c46e88dbbec62b3178af1 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c89f2fa2824925a509055404a63611 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2691d858462b431185fa7b153a2bd4c7 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78618dc91c264b1a933f96a13b1a8725 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f82a16cd78a41cbb749df20915c8e82 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c181e9ffe32b43a69d5be122c8b56233 : SUCCESS in 19m 20s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0214c3e29a8b4ba39d2510179835dd71 : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/758edd3f59974d66bde133b528e7c0e1 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a38b2810643148d5a12492ac7050f7be : SUCCESS in 16m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/96eb5c9085af4f849d56f3d36d3fc826 : SUCCESS in 20m 04s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6a0dec929a404108b6bbc6659ea2abf2 : SUCCESS in 2h 08m 59s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/85e8c5a338cb481d8d5a9c86ecde90af : SUCCESS in 22m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4fd067a8f8f2438abd1634699df8f807 : SUCCESS in 23m 36s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/31c99dff05a64000aeff33cc5cf1cf2c : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"08cf5774e9cfdcb6e7ce2abcb15091bdd6a7a284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:02:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21bc45ccebdbbe85c6545ee2fa4d19c72710a26c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:04:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05037174dec84611911236a872b9cb46 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ef66f65caf74050b00e1b5e800e6c65 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"244bb221cdd918f2646fa99ffce4bf01e29a479b","revisions":{"244bb221cdd918f2646fa99ffce4bf01e29a479b":{"kind":"REWORK","_number":1,"created":"2020-09-17 13:51:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/752472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/752472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/752472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/752472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/752472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/752472/1"}}},"commit":{"parents":[{"commit":"739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac","subject":"Move dnf.yml earlier in host configure workflow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a2412f9c5c18bcd45deb8d73ccb3561f6e9ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-16 13:28:54.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-17 13:51:50.000000000","tz":0},"subject":"Avoid setting IPADDR\u003d0.0.0.0","message":"Avoid setting IPADDR\u003d0.0.0.0\n\nConfiguring network interfaces with IPADDR\u003d0.0.0.0 may fail with:\nError, some other host (FF:FF:FF:FF:FF:FF) already uses address 0.0.0.0.\n\nChange-Id: I908dc3320cb1dce7a040a1dbfeac414ace22d6e3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nStory: 2007900\nTask: 40289\n(cherry picked from commit 2606c23812be510c1f8ced03a5e0c32a218d827d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/244bb221cdd918f2646fa99ffce4bf01e29a479b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/244bb221cdd918f2646fa99ffce4bf01e29a479b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
