)]}'
{"id":"openstack%2Fkayobe~752686","triplet_id":"openstack%2Fkayobe~master~I9df2f98772c143a9662fd5c123cc9f6d5410c5e5","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I9df2f98772c143a9662fd5c123cc9f6d5410c5e5","subject":"Remove unnecessary requirements from tox.ini","status":"MERGED","created":"2020-09-18 13:03:25.000000000","updated":"2020-09-21 09:23:57.000000000","submitted":"2020-09-21 09:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752686-1600680067261-c6eda660","meta_rev_id":"0d951f131ae1270464c1725f44ffdc416ed51738","_number":752686,"virtual_id_number":752686,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 09:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-18 13:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 08:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-21 08:55:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 13:27:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-21 08:55:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-21 09:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5f5206cc24f21557a96845a3db5ea3f7f19aa53","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-18 13:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a226b1ff38a0d2a35101682202ded0aac86d69ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-18 13:10:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cfec8c22c5d3925b7b8dffe9f1610f5eda1b3dd9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-18 13:27:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"85babefe655ea465a55421b5630822ece32c7293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:39:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc65030a45664af1bf14b939523bf5c5 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c926bcb07544d7a1fc27f23ca7a36d : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c8ea1a1427445ce8f7a3c2d0f567a0d : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35f924725ff54e09be4d91a8c2f6bde5 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252c225f774a4703a7a6020ee2ad294f : SUCCESS in 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32e0113d03c343418610b0b0f3c8e718 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/055c03424210498cb5b5f3e3d7674168 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb04db257c0d4705be2213c3418d6f04 : SUCCESS in 7m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/99c80b1d1eda4b928bc96097424684a9 : SUCCESS in 21m 33s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"a4969763f68b606e781d308bf1f4af0c2846426d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-21 08:55:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"214efd32311b74a554e107ff17e18b50449378f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:55:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b5b42d0d5e280bc9b779a73950eb8b1e970f453","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:21:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"57dbe93889396b9c878902924caed99b848bf09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:21:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b326532b5274064a10e3dde9767a7cf : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3feecde4e1c4a838f828ad4590b06c4 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce2d5ecb19554b03bc2db67fc1efe666 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc9196c4ef44fd18853c30168cee616 : SUCCESS in 17m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/603c1371a35e491fa7116fc729ed9764 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b8e0a1ae3f0444aaaa6d35bba940857 : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3bec105b59404e8c981f9fd570d2028a : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5065b9af188d40d28c64a602f26a105a : SUCCESS in 23m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"0d951f131ae1270464c1725f44ffdc416ed51738","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5ea3ccf5f74c34b577f9189cb8dc40 : SUCCESS in 1m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4286aa627954495a288a8a2dab92012 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9475050cc7227ecdb600139e832839686dfae2a2","revisions":{"8d86ca776e2d675639dc203bb6d78388a330be3c":{"kind":"REWORK","_number":1,"created":"2020-09-18 13:03:25.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/86/752686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/752686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/752686/1"}}},"commit":{"parents":[{"commit":"79426d074025f50299b1db5f78788a60abad3175","subject":"Merge \"Support neutron-mlnx-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79426d074025f50299b1db5f78788a60abad3175"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-18 12:49:09.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-18 12:49:09.000000000","tz":480},"subject":"remove unnecessary requirements from toxi.ini","message":"remove unnecessary requirements from toxi.ini\n\nrequirements.txt is unnecessary when run tox -e releasenotes,\nreleasenotes relative stuffs is in doc/requirements.txt, this ps to\nremove the requirements.txt from tox.ini and reuse [docs] deps\n\nsync from kolla-ansible and kolla\n\nChange-Id: I9df2f98772c143a9662fd5c123cc9f6d5410c5e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d86ca776e2d675639dc203bb6d78388a330be3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d86ca776e2d675639dc203bb6d78388a330be3c"}]},"branch":"refs/heads/master"},"9475050cc7227ecdb600139e832839686dfae2a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-18 13:10:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/752686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/752686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/752686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/752686/2"}}},"commit":{"parents":[{"commit":"79426d074025f50299b1db5f78788a60abad3175","subject":"Merge \"Support neutron-mlnx-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79426d074025f50299b1db5f78788a60abad3175"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-18 12:49:09.000000000","tz":480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-18 13:10:43.000000000","tz":0},"subject":"Remove unnecessary requirements from tox.ini","message":"Remove unnecessary requirements from tox.ini\n\nrequirements.txt is unnecessary when running `tox -e releasenotes` since\nrelease notes requirements are in doc/requirements.txt. This PS removes\nthe requirements.txt from tox.ini and reuses [docs] deps.\n\nSync from kolla-ansible and kolla.\n\nChange-Id: I9df2f98772c143a9662fd5c123cc9f6d5410c5e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9475050cc7227ecdb600139e832839686dfae2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9475050cc7227ecdb600139e832839686dfae2a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
