)]}'
{"id":"openstack%2Fkayobe~752900","triplet_id":"openstack%2Fkayobe~master~I3014983f481a5dca7c93e140b3e10caa5d537669","project":"openstack/kayobe","branch":"master","topic":"barbican_conf","hashtags":[],"change_id":"I3014983f481a5dca7c93e140b3e10caa5d537669","subject":"Add missing barbican.conf support","status":"MERGED","created":"2020-09-21 09:01:09.000000000","updated":"2020-09-24 10:31:01.000000000","submitted":"2020-09-24 10:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752900-1600943332208-e4987785","meta_rev_id":"51fb37c06103e2c1e5806869d8c2f42bbc62d6db","_number":752900,"virtual_id_number":752900,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 10:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-24 08:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:17:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-09-23 19:17:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-24 08:10:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 13:29:13.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-23 10:28:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-23 11:25:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-09-24 08:10:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-24 10:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c252c1d9dd63808c44fc5e9cdb212d4fb95ab526","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-21 09:01:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715b68f1a579aae63cf2d3af3432085887f4e6e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-21 09:04:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f456438745a4d4e49ad57631a2ac68fb751d6984","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-21 09:10:42.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good.","accounts_in_message":[],"_revision_number":2},{"id":"57aaaa6985c2958ce0b42882237c9bb12fe67cf3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-21 09:11:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cea31af04661dd2405c474d13016b0f918d1f2a7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-21 09:11:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9916693a2cd86d969fdb64a18e1920a807b39d50","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-21 09:13:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"69998e38f9e4728c39b1824c382eea09ef612ee2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 09:38:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"609879d3f0d7cfc513117313ca656cfed973bb12","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-21 09:48:28.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"c0905c9e30767b555ab43bbff0e91346d49f5445","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-21 09:54:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8ab25852c71ff0ca01de55966bc1622d675687af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:51:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e289f6357c47d381009f65821b83dd : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa39361781674779b3bc3b061f2dd8c7 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9df1541f28674f48ad5b3399ac63a582 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3416a30838084471bf0bff06d4309746 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f264d65df240ce97810ccc4e27a533 : SUCCESS in 14m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1fda0993f54345698d75c4dcd0793abf : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a52266b8e354776894f16092f9dd996 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c57a235e24e74eae81d9d9805f8e6ad0 : SUCCESS in 22m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cded96161fb647e3b6176e269739fd22 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/26a8e17bbd6b49999a806d120aaede17 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2e4937f4179f4e69a9826494a0fee7c4 : SUCCESS in 1h 49m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/70b842f69cdf490b96f3d301b17f070f : SUCCESS in 18m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e44751b6a93349628b5a0a598a3fe8b6 : SUCCESS in 25m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/20f6734eda9c451db86d8b19fb303012 : SUCCESS in 12m 09s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"303f3993b9351eb1140dcce01ef0b0ac9d8ee37f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 13:20:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c7760c1c19d5fa4b7940dedce9b5272067c4ccba","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-21 13:22:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"eceb483599eb04cad15b9d95d565bac46b629935","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-21 13:29:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"69f82e05acc0038154a9ab5706ac9baf644d7a08","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-21 13:57:18.000000000","message":"Patch Set 4: Code-Review-1\n\nShouldn\u0027t barbican.conf be added to ansible/roles/kolla-openstack/molecule/enable-everything/tests/test_default.py?","accounts_in_message":[],"_revision_number":4},{"id":"37471b67903191ffe0d40870007a3a6ee492dc9d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-22 05:13:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7b2dc8be6ce1d18e7e3647a1aab5041c6e6c6d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 06:55:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c62ab68ecd040dd8ff56f321ddf2557 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ec0076fb1f49c3a593221a18d6469e : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f37199cc8b994b7ea4e42e42ec7ef438 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f28bcc0e67b244a1bcd980e15ee4ff83 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7da799a761b4be0a1a785f673204d77 : SUCCESS in 16m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4be51cb498ce48caabe3de57a4577a3a : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6246a36b76234de79140fe5bb5142fd4 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bda2df0ba1494243bf0fb52da4418a5e : SUCCESS in 19m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2f6a75882b714dd085f911f8daee2799 : SUCCESS in 59m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e220afafd6444b44941926262585dcea : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db899842734f405abcac167792f68013 : SUCCESS in 1h 39m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/52128e5ba5a44c3792afd79c7a0c6c41 : SUCCESS in 16m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ba78500f2462455680f5a87339716be2 : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/beccefa581654188bd3382df2d45dc04 : SUCCESS in 11m 25s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"0d46323770e5c660a16a417443177e2bf4c3ef6c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-23 10:27:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5313a9e7cb07df31d30dac7ae6f9e306233aa5d3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-23 10:28:27.000000000","message":"Patch Set 6: Code-Review+2\n\nAdded a release note.","accounts_in_message":[],"_revision_number":6},{"id":"7635868b62fd69148327f5645e9cb4a1b8a6cfd5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-09-23 11:25:12.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b869b354912fe2c35457bee27a8c74517a87b81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 12:07:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c6bf2305df94b35b769a0d6fd68697c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a281819c5d742aca024869f933df99c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a87af9f4dbd041329838a67d6b9785b6 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/257dca242a5b47adb52393a85ea857bb : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a922d1b1628f4c78b5cbd2fa7fe41608 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b993ee25e994d02a0b372ca506ffaab : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23b4cfac5fe642a7963e852cf75deea5 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddac27787a67474a9b32d1914adc3494 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3512f023f5344078e921bb1ddef08a6 : FAILURE in 7m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3278e101498a4060b576398639f972f1 : SUCCESS in 56m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b626f88a0f474ba1b4e0be610775f3ce : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ac6e3d1e2e0e4e75b7f3a744abcd4854 : SUCCESS in 1h 37m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6027e141dbc04d4b83e40d920ac38ead : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/37023e174a574337b9d33c8a2f4ab3c9 : SUCCESS in 23m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d08af6d6ac2844499b1b435291b07d2b : SUCCESS in 11m 44s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"f47b01f6c2a657153bd5a33a4aebdc3d7d1604ac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-23 15:31:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8644e1e759353ce719694b6b83b2c311fa6bd025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 17:30:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4870c56f4a1643e2a232412679fa1137 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72b2fcbe7ea46e0ae7865224b0719ed : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b64faed26544a1082674bba4e7b72fc : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d245194a8b3344f4b0eecad1420db862 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36aa16f468c74bfdab2b2a421b126f88 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94d526b4c4da486c832ab1bbee392a0c : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a96723a4c2242bf855dc9b24917e6aa : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7dabc65c6ae4408b39b4d272b95e319 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04dff04afb2c477fb8a2cdc2721160e5 : FAILURE in 10m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/42b1ffb64bc246e7893770e2ac72aa09 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ac322b38383c496295d4f2a043504fb2 : SUCCESS in 22m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/669ca169d5ce4710918d561dc1c1e89f : SUCCESS in 1h 48m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/78e61698dc0a441baf6fbfdc3615e674 : SUCCESS in 25m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af9411ce1b3343b8aabe394c90f403e4 : SUCCESS in 34m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/654c354b63b742c7bf669e39d06b0da6 : SUCCESS in 16m 15s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"9f81131a4a278d2137d9f990abd77122ba1826d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-23 19:17:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"4be4b654e87bc5ab5036f22b6c0d0bc5efa4c6d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 21:14:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2089e522a3b649a8b980d8972b4ab68b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387fe3c378974b498192ac6a58020ecd : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b9d717bdde49b5aeaec975a2f948c3 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/316d3bd8add14041ade9c1d8bebebfbc : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901a77d6a7854b569efa26077f51f9ea : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ceb61a355d348d48f7107038e777f10 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c72f3d58a9c47bba37bb271b9399139 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9afe6bf01870433abd442784dc09b8de : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4bf2b8fda57e425fae235fd0071f2f9b : SUCCESS in 20m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6989a325b737412397911736e3ff1513 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4bb5bf8cb9cb44b793e3061e0aba3d7b : SUCCESS in 16m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/402182b3e5f54860b8e8c01e08dc4dd2 : SUCCESS in 1h 52m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/911e427552e0408bb6aa7fde664a92ad : SUCCESS in 16m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b3bf6872dd964a659eb99758ae3ae2b7 : SUCCESS in 28m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bb52e456c6ac42f2b8c77bd81b0eca39 : SUCCESS in 9m 42s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"52f89c8a25209e8259c1587cdaf3e5d58d511a2a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 08:10:14.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b8afdcff64c0f5b654c236f8b1c7ac131cd0bf94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 08:10:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9537196c2d311b08d031e1d9c31a66e4e676d562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:28:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79b051fa705e492ebafdc24a928d9dfb : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8723117ec1d54ee68a8bf211a97f15c0 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3cd0b04f6cc49bbb2c49a1628a7f3b7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a8aac26ee54e8eae66cb5872d09f05 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65bd0a05d3494bcfa99bed73d8335371 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2025cd24f71947b5a38f960f7111811d : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/518c9da5b0094ea5bccdd502c9262498 : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a667da9b40b4a309d6bfa5a208c383f : SUCCESS in 21m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/16e23233b06a4d32903db9e52811f704 : SUCCESS in 54m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/527552ab3fad48c8a3fc6390c83c4d80 : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/625da8d07c0044f387d2f7cc662ec7ad : SUCCESS in 2h 10m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/07614799fe8d44d4998045abb5ca55b1 : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c8d5b87a07f40818c8fe1740cd32a23 : SUCCESS in 25m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/13fa5c0a21a645c9ab91aee8c5dd603a : SUCCESS in 11m 18s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"b64ed61007c388285e8f3361ac7510956e53191c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"918667e1e265ca9bc126aca30badf92920f9d0e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:31:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9468261b217b45e1b01df2b1ca58018d : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b5c912b7c164b7d9c5f48226438857e : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0f25900a43895f5154820292d44ec999cd1ec2ed","revisions":{"0fe6c9af4f925b04cc8280fbf382dcd8b0f15bb0":{"kind":"REWORK","_number":1,"created":"2020-09-21 09:01:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/752900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/1"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:01:06.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe6c9af4f925b04cc8280fbf382dcd8b0f15bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe6c9af4f925b04cc8280fbf382dcd8b0f15bb0"}]},"branch":"refs/heads/master"},"b448a50f6408c4796e1649fda94f8b8eb89fd7a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-21 09:04:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/752900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/2"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:03:34.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b448a50f6408c4796e1649fda94f8b8eb89fd7a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b448a50f6408c4796e1649fda94f8b8eb89fd7a2"}]},"branch":"refs/heads/master"},"9292b6d5fcb03580466013e21db8725640d9d65d":{"kind":"REWORK","_number":3,"created":"2020-09-21 09:11:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/752900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/3"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:11:18.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9292b6d5fcb03580466013e21db8725640d9d65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9292b6d5fcb03580466013e21db8725640d9d65d"}]},"branch":"refs/heads/master"},"7be4d3b13b51042c2b8e626190f4e8ac9448bbb5":{"kind":"REWORK","_number":4,"created":"2020-09-21 09:54:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/752900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/4"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:54:32.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7be4d3b13b51042c2b8e626190f4e8ac9448bbb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7be4d3b13b51042c2b8e626190f4e8ac9448bbb5"}]},"branch":"refs/heads/master"},"f6be6d481d7f54f00a9018d41b200e803dcb2fdb":{"kind":"REWORK","_number":5,"created":"2020-09-22 05:13:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/752900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/5"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-22 05:13:22.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6be6d481d7f54f00a9018d41b200e803dcb2fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6be6d481d7f54f00a9018d41b200e803dcb2fdb"}]},"branch":"refs/heads/master"},"6b7cb002070e9842311cdf51e3447270ed405f80":{"kind":"REWORK","_number":6,"created":"2020-09-23 10:27:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/752900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/6"}}},"commit":{"parents":[{"commit":"b778ac01cfcd0f2f520ca8b76d97b15c367f6a59","subject":"Bump hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b778ac01cfcd0f2f520ca8b76d97b15c367f6a59"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-23 10:22:38.000000000","tz":120},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b7cb002070e9842311cdf51e3447270ed405f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b7cb002070e9842311cdf51e3447270ed405f80"}]},"branch":"refs/heads/master"},"0f25900a43895f5154820292d44ec999cd1ec2ed":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-23 19:17:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/752900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/752900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/752900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/752900/7"}}},"commit":{"parents":[{"commit":"c58555498836066002dc33620018e9bd1e2eeb6a","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c58555498836066002dc33620018e9bd1e2eeb6a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-09-21 09:00:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-23 19:17:51.000000000","tz":0},"subject":"Add missing barbican.conf support","message":"Add missing barbican.conf support\n\nStory: 2008170\nTask: 40925\n\nChange-Id: I3014983f481a5dca7c93e140b3e10caa5d537669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f25900a43895f5154820292d44ec999cd1ec2ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f25900a43895f5154820292d44ec999cd1ec2ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
