)]}'
{"id":"openstack%2Fkayobe~752955","triplet_id":"openstack%2Fkayobe~master~I6946b578f23843dd3cbafcb185b8d8cfbb00c080","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-07 01:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6946b578f23843dd3cbafcb185b8d8cfbb00c080","subject":"Fix seed VM configdrive when admin network is a VLAN","status":"MERGED","created":"2020-09-21 13:42:31.000000000","updated":"2022-01-07 01:56:24.000000000","submitted":"2022-01-07 01:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752955","meta_rev_id":"5b990043760cffefe9f41ad0dda80736713c256a","_number":752955,"virtual_id_number":752955,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 01:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-30 11:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-06 13:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-06 13:49:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 15:32:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-30 11:56:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-06 13:49:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"4cfd3e544455f030e61126eb440b797c75d70fab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 13:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a01490ef9a90a0fc36f2b5b1cf1e1ba4c0335e6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-21 13:43:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"185b3e951d5365a6fb06668e2743755be3f28a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 15:32:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f44753a89614794bf1988daff761a66 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0897d4e4b23e48edb4bc069d5e26085e : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80247a0e2b264ddca7984fb416416afc : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de4cfcd4f5a549d6aed03630a912949a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a70bcc91f504b53aee0f74ba903e0c4 : SUCCESS in 19m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17757c23ab864392a908c68c49202458 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aaa588c1528e43c3af93c38c1416997a : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aee0b0105a8e4d1286f7ee2edcb03480 : SUCCESS in 21m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17f188fdb02b48638a3a50277bd25b28 : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a579ef07be7e49e49990a761bc68513a : SUCCESS in 18m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b4269c7680684dec8b5b62fc6d9f3e41 : SUCCESS in 1h 43m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/64cbc8272473455381e3cc31679ce1c3 : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7715062ea18d4a2182aba2c94b4fbda5 : SUCCESS in 31m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6e0612f581cd46f38aeffcebef088592 : SUCCESS in 10m 09s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5d1b7a320f864d3771b7f1e3fe3eeff4839b0f36","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-28 08:54:53.000000000","message":"Patch Set 1:\n\nAccording to Damian Dąbrowski, this fails in cloud-init during boot:\n\n[ESC[0;32m  OK  ESC[0m] Started Hostname Service.\n[   17.017284] cloud-init[794]: Cloud-init v. 18.5 running \u0027init-local\u0027 at Tue, 22 Sep 2020 22:19:33 +0000. Up 16.45 seconds.\n[   17.019197] cloud-init[794]: 2020-09-22 22:19:34,192 - util.py[WARNING]: failed stage init-local\n[   17.024692] cloud-init[794]: failed run of stage init-local\n[   17.025708] cloud-init[794]: ------------------------------------------------------------\n[   17.027133] cloud-init[794]: Traceback (most recent call last):\n[   17.028166] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/cmd/main.py\", line 653, in status_wrapper\n[   17.029939] cloud-init[794]:     ret \u003d functor(name, args)\n[   17.030926] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/cmd/main.py\", line 362, in main_init\n[   17.032644] cloud-init[794]:     init.apply_network_config(bring_up\u003dbool(mode !\u003d sources.DSMODE_LOCAL))\n[   17.034179] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/stages.py\", line 649, in apply_network_config\n[   17.035989] cloud-init[794]:     netcfg, src \u003d self._find_networking_config()\n[   17.037220] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/stages.py\", line 636, in _find_networking_config\n[   17.039113] cloud-init[794]:     if self.datasource and hasattr(self.datasource, \u0027network_config\u0027):\n[   17.040611] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/sources/DataSourceConfigDrive.py\", line 155, in network_config\n[   17.042664] cloud-init[794]:     self.network_json, known_macs\u003dself.known_macs)\n[   17.043914] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/sources/helpers/openstack.py\", line 632, in convert_net_json\n[   17.045948] cloud-init[794]:     name \u003d \"%s.%s\" % (link[\u0027vlan_link\u0027], link[\u0027vlan_id\u0027])\n[   17.047300] cloud-init[794]: KeyError: \u0027vlan_link\u0027\n[   17.048147] cloud-init[794]: ------------------------------------------------------------\n[ESC[0;1;31mFAILEDESC[0m] Failed to start Initial cloud-init job (pre-networking).","accounts_in_message":[],"_revision_number":1},{"id":"42609801a00f938aec4a1e3cc5ce14a29b3b0fc2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-09 11:28:09.000000000","message":"Patch Set 1:\n\nThis does seem to work with a CentOS 7 VM. Based on the python3.6 in the path, it seems Damian was using CentOS 8.","accounts_in_message":[],"_revision_number":1},{"id":"43565a404ac9fa4462e06728b714dc84ffadb9c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-17 14:04:11.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e According to Damian Dąbrowski, this fails in cloud-init during boot:\n\u003e \n\u003e [ESC[0;32m  OK  ESC[0m] Started Hostname Service.\n\u003e [   17.017284] cloud-init[794]: Cloud-init v. 18.5 running \u0027init-local\u0027 at Tue, 22 Sep 2020 22:19:33 +0000. Up 16.45 seconds.\n\u003e [   17.019197] cloud-init[794]: 2020-09-22 22:19:34,192 - util.py[WARNING]: failed stage init-local\n\u003e [   17.024692] cloud-init[794]: failed run of stage init-local\n\u003e [   17.025708] cloud-init[794]: ------------------------------------------------------------\n\u003e [   17.027133] cloud-init[794]: Traceback (most recent call last):\n\u003e [   17.028166] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/cmd/main.py\", line 653, in status_wrapper\n\u003e [   17.029939] cloud-init[794]:     ret \u003d functor(name, args)\n\u003e [   17.030926] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/cmd/main.py\", line 362, in main_init\n\u003e [   17.032644] cloud-init[794]:     init.apply_network_config(bring_up\u003dbool(mode !\u003d sources.DSMODE_LOCAL))\n\u003e [   17.034179] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/stages.py\", line 649, in apply_network_config\n\u003e [   17.035989] cloud-init[794]:     netcfg, src \u003d self._find_networking_config()\n\u003e [   17.037220] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/stages.py\", line 636, in _find_networking_config\n\u003e [   17.039113] cloud-init[794]:     if self.datasource and hasattr(self.datasource, \u0027network_config\u0027):\n\u003e [   17.040611] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/sources/DataSourceConfigDrive.py\", line 155, in network_config\n\u003e [   17.042664] cloud-init[794]:     self.network_json, known_macs\u003dself.known_macs)\n\u003e [   17.043914] cloud-init[794]:   File \"/usr/lib/python3.6/site-packages/cloudinit/sources/helpers/openstack.py\", line 632, in convert_net_json\n\u003e [   17.045948] cloud-init[794]:     name \u003d \"%s.%s\" % (link[\u0027vlan_link\u0027], link[\u0027vlan_id\u0027])\n\u003e [   17.047300] cloud-init[794]: KeyError: \u0027vlan_link\u0027\n\u003e [   17.048147] cloud-init[794]: ------------------------------------------------------------\n\u003e [ESC[0;1;31mFAILEDESC[0m] Failed to start Initial cloud-init job (pre-networking).\n\nReproduced on CentOS 8.","accounts_in_message":[],"_revision_number":1},{"id":"221844905d5232b610dab15ffb16204cbd377020","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-17 15:45:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b22dffe4c7e6e413f46adc180d965d9572410769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:31:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d787dc2318b640bba605a32b4fb1b8ae : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9a0b01366d46c7a3d06a211baafc3c : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7b43a941c0c4959a5f1ed3e0aa786bb : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66188c48f0a14af2844e5da26933f1ec : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/892d78266fff443397a356f2b769fc4c : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7662654a7d4a45f4bbc0e4a08e9ace39 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79832b2a16d402bacb33cca050d11c1 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a6b975e2241412a9ee191c01d63b113 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/616f1817c3b04902acab0428150712f9 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ed6d9b16f014d56848edbd6117bbae7 : FAILURE in 8m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5900aeb632ff44a986cc80b524c194b1 : SUCCESS in 58m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b19d42f4bd24956ac2b008ce7e08569 : SUCCESS in 52m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/29175cbe9fad4d37b083c341f953d02a : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/03392d52c02f44c7af5beb1e03072ce2 : SUCCESS in 18m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76b784228af24faf9f6d74a64d5ba450 : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c61b9c14e6a849d3931e5da354915579 : SUCCESS in 1h 41m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84a3d23372e44d1d96661d29aae33545 : SUCCESS in 34m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76892ff993d3496091d01fd3ee82dce9 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ac0484caf3d04f85a8d2bd22cb4382ed : SUCCESS in 25m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a15b2dd5cbe4876bb89c07060aeffbb : FAILURE in 11m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18ce5ef3107548c5ba14bc851a5e0b60 : FAILURE in 11m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"e280cad7fcf616951bd37ada38433e471979c224","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 08:18:24.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2d680c46f83f6f6f2d2dc77ec8ccbb0aecea23ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:41:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cbdad6f69da4eba99b2ed8e3adf4b94 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5b43bffbe241678bfc6b993029515b : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e063dbacdc44855b0bc24e366eba7c0 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bacfe2b0d176467dbd587093e3a2305d : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/236ee38fd24c46baa3aa34bc65f64b6b : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e40bba3d44494499fc5ce4869e82f8 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6345cc47eb834d0eaddf3197eed85d34 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70ab32d7ed4945ae920cffed7d594692 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7bd7956800e420ba14779ff2b574df4 : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/043b285f6d354d2cb4b972e12708ed61 : SUCCESS in 30m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3eef91b341bf4ff7a7ce0e24aa387314 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79c8bce199b440fa3d5061bc37c3e63 : SUCCESS in 46m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bc8f2be689044d37a5a780903753ddac : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a6ccc0a0cf44bd89645557270ee05b6 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf00a717c314bcc863dd1a34ef7c58b : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/691ddf7ae7d6454c8b8822c120b97143 : FAILURE in 1h 21m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/78dcdcdaa76940929bccac5787a29557 : SUCCESS in 37m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55fd832b185d430bb1e9c8a4e9aa4189 : SUCCESS in 25m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d9138ea22d3e4747bbf672b1bd2ec79a : SUCCESS in 29m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/83e60a54d6ba45f0a7f1b9734c33f3d5 : FAILURE in 12m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a87928210e04828a8b43a9eafe3610c : FAILURE in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"eaa662e966cf1fe314e9ac9d39716831b0674157","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-27 12:02:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e6b0aca48197aeace7548b46264b42f8101d3775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:53:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc33d832beeb43b38d49c63f8319fa3f : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a7e67574a246d7899569f4d80e61c1 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eaa7d6612624f7e9ab6282ccf48d17c : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6a3e3dd2a34f03b5114b8bae365894 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/934efb2a42744d9cbbee9539ad3f7ec9 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66c29288f3844f28bfcb60752618675 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a474200e17945069822ffe38ec974ae : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed48e0a156ba4e6fa2bf439733d898d4 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5a76f2bd7d84e83b294fe8f9a2f2328 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79f042bf258f4c738cc0e06cb32cf652 : SUCCESS in 30m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d0e0bf58556949e88cc0daae239286bb : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0552a39974ef4a62b4b033b0b97244a7 : SUCCESS in 48m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8dbb6cc635de4f639b94ce626873be34 : SUCCESS in 1h 13m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d321ee77e6444c385e990e7defde189 : SUCCESS in 19m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/427b3bfe97334dc4b5be3d2a3d282946 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/17382daae0394bfebcca9551c14f61f5 : SUCCESS in 1h 49m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5e22241789d94b5290fd98347132db05 : SUCCESS in 37m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/974d17076d85425eb6a31afcfde1c1b9 : SUCCESS in 30m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8625b90cae344c4385ac96c223c05558 : SUCCESS in 27m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/593ecfde1e3e4d3da4d7cc176d1f1af5 : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43be67a2635e437ca64623db78b285ea : SUCCESS in 12m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"74ea5d22fb9b3cad9d276701e2f9f659e756aad7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-05 14:43:55.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"855099898d055f89c91ea9bb44feb2eccd25d49e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:44:23.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"2e8be6f6dae3a9c865781b38f42f457908ad22b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-05 14:45:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a892163884b186666e11fbc3dae902809b07e01c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-05 14:45:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"74813a23d27a59cdbfd0bef43de2f22bf055490c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 16:42:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a4acc09cbd14d7bbf123fcf9e98ec68 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e60fb35add94ad0b58122c70e6c5e1d : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e329667eaf2a4d36a1359261ee226234 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecc9bc6dd314467b45f02c168325773 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da90f21ddf748e1b470f222719030e8 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bde977f73c64bb09d7a9c7b6bf3f405 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c325ba14b04747299c6195ddafa3d90a : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/161f9327d8174f62abfa39411fd02df1 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bc0b5b7bad34b9c9bfaef34c3ffcdd5 : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5623128465942babfc0b6627a1d9c13 : SUCCESS in 34m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1a4da37d1cd04f66afcf60bfb3d0f3dd : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3fa95031099420285ebf68831eebaaf : SUCCESS in 54m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/df7f4de2f9384a13abdf4e44aeca4b88 : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/745c7a2493d247d8b04ba24c5a7416f2 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca004b004b05474db9d403d756d4a269 : SUCCESS in 26m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3145009b1f2848e393a3233e096aa67b : SUCCESS in 1h 53m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b52fafe92544242a0dc1877bb104b4a : SUCCESS in 1h 37m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1b4f5e2ec93c4ac78e4a8d949806b8ae : SUCCESS in 40m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d26db4a41454b71904630a1c512ab4d : SUCCESS in 38m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db0ed41f0dd143ff9f5fc14bb1e86fda : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63cc40efd94e4194bc029ce73709899d : SUCCESS in 30m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8987b281e4e5435e9aa51a027ad4c1f8 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7306dc81f416471fb8e1fe4cf9d3ae67 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":6},{"id":"075bf6dde57a372f3323331ffc5023a228dd6270","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-30 11:56:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9af360eeee8c631b768677c9d89aa4f8cb35822c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 13:49:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7f561ee22a9e35af651265d09e5e8459cf474415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:51:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"470cb0737fcc4f33f9fbe9bae5937f328b07b15c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:47:19.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ae4163ed3446508517983e9503546a : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b849821f8d24895a3867b66384572d9 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a22d76b5c68649d39eebe8b4a94af32c : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a79b783917941acbf956d9a8e1cdce0 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f512a9b3c7f14991978b11243e9c4e96 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a80667813a72410783bd8f99c82f40b4 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7065a2faecb9414e92e70bfa890a613b : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f24e119c7e154d2b96e531610fd21aa3 : SUCCESS in 31m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b68e3d5fbfe46d583e5af1f5c13ebc9 : SUCCESS in 1h 17m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd787dbe0a934e7bada4494deb5ece66 : SUCCESS in 44m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b45a4c044149477e8274b52f02aa6f5e : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f3598f91c4f14d06aa7cf6ed14619e74 : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8800c8f5dea646cea169e9f2957dc832 : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a9d0fd4e44f4742ae1157ee2dae3761 : TIMED_OUT in 2h 42m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d02bea047164629a49c1aac80d2fe5a : SUCCESS in 1h 24m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e98f108058574a659e98a1bce734f254 : SUCCESS in 54m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78e78eafc693435fa025f615b91a9b8d : SUCCESS in 42m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f73e75b9deda4f5bbb3eda877f3cd5db : SUCCESS in 30m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20a6e4f739f4217bec29247c0eb3572 : SUCCESS in 25m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/06c6beac5e2149989272d72995278ae0 : SUCCESS in 17m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7429dde29f21432e9507e937ef69e4c3 : SUCCESS in 14m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c22433c92b064bc18e8e86711e6d195b : SUCCESS in 19m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23ac4b093281425a9711db9ee07f73d7 : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a1e78de05f099571efc268d511c349fb8d20a55b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 22:12:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77430fb4787341ef17218b0f9d514de9fa5e2389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 00:02:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0edeba91faf465490cdf009e842fe8b : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2630c82069846c4bb919c06043e8977 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66695dff2a654962940d8cfdc2966c92 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a71bb39aa7040f1ae453c89f563b045 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e75ea1717ba4b63b041cb3ee4879fbf : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8015ffa083c443458384080ae6a83bc3 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff5d274485f54a3ba48ebcf933728387 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c07aed1bb0144697aad8ccae23861241 : SUCCESS in 14m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/419be083736245b5a2005c574555966b : SUCCESS in 32m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7edab9175904194850f277f14e1f9c3 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7baceb9c2a174b8195467fcaebc440be : SUCCESS in 45m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ceb0a9af746d48e6ad5fcdf76eb40419 : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/585f668b7bff40e78f274020706dc76f : SUCCESS in 25m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39287ac92bf644d28b5712bd017a9afb : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/10405dcd725b440eae3cf9050c05c4f8 : SUCCESS in 1h 48m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/985037fb2775457fa4de9c75e881f3eb : SUCCESS in 1h 16m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/00cf2aed556b4e049e10a5f77d26b50d : SUCCESS in 49m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/222e1c622ee34f5fb5e60ef3562690eb : SUCCESS in 50m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d68487f091384beead5db5ffec64c1f0 : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d95a52d951bb4ef58da8db1ede1c8a24 : SUCCESS in 23m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5999103cd2a43c6bf7687e349f10d5c : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ec154b386d41adb15daf6e1344b0bb : SUCCESS in 11m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed41fcef967b4d5f9a80c39238c7354b : SUCCESS in 18m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0125e56998c41ebbe05497e2e687572 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":6},{"id":"17245743b08f898a87d9b75eaf7f47effc4d1155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 00:03:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"45247187ce56e2c2afc4e3d096c6c5d92acaf620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 01:54:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ed3028c04644cd99ef378bda663c78 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3df263aad4a54c3387bf6209e44a08e4 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2069e026690c403f9f372c3b2962729a : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf9cee91ed746a6ba5a40ab43fe7748 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e63d71042224ad493e5d866bc4dd927 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6c1a03eedd44133aaf8b06471055343 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a4206b58a2f4c1a8159c0dbdf7f8e9b : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/676767ff245146499cdd48571e3ef157 : SUCCESS in 31m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3f95d8efba8f47ed8726616b99a976d6 : SUCCESS in 1h 13m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c65eb4db1db744df9f4185589527c87f : SUCCESS in 49m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f22bfa53a2ea408facf42c664b473a09 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7e50d326c76b4b6994f8646cd46ef960 : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41de5440e472406e9feec5b56ec2759a : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/97df74f2371948c5b8b7cc07ac213a4e : SUCCESS in 1h 47m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/922a9879218949cd9a84db99f2ccf4bf : SUCCESS in 1h 06m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b4a94c2550804a90bc05f2f90fe039e4 : SUCCESS in 1h 05m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9ac13260475499492b4e0e79489d295 : SUCCESS in 41m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/119e792bef6a4fa7a3d519460a51bc02 : SUCCESS in 28m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d02a98051594b259b74e7db838beeeb : SUCCESS in 21m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f481503744a47ca8ca54b2334c693e3 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3205d880c6ac499198ba810746964af7 : SUCCESS in 12m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0875bd72b8824df7b602d4e676adeba3 : SUCCESS in 21m 03s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5dcea00b12a4de9a87fa8f57fc236cf : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":6},{"id":"88e414d9fa7552c7fc9c1369cd484dc999efdd1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 01:54:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"5b990043760cffefe9f41ad0dda80736713c256a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 01:56:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d3d2b22bce4a69855bb0bfd811de6b : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/066fa7f26f73424c93ca7431300cc13b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bc0885f5397af9ee69f8c873e57ad6f3d7a360e1","revisions":{"fbb3dd14da0ea2cd9997e24a78975405e4e05a76":{"kind":"REWORK","_number":1,"created":"2020-09-21 13:42:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/752955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/1"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"subject":"Pass type\u003dvlan to seed VM configdrive","message":"Pass type\u003dvlan to seed VM configdrive\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbb3dd14da0ea2cd9997e24a78975405e4e05a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbb3dd14da0ea2cd9997e24a78975405e4e05a76"}]},"branch":"refs/heads/master"},"72c4bd66488c4c14d6bb018f57d18dc1bdd4ef35":{"kind":"REWORK","_number":2,"created":"2021-09-17 15:45:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/752955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/2"}}},"commit":{"parents":[{"commit":"f1212d0f07cd127c4a618126178a9edfb44160e6","subject":"Merge \"Support setting docker registry credentials to the empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1212d0f07cd127c4a618126178a9edfb44160e6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-17 15:44:54.000000000","tz":60},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72c4bd66488c4c14d6bb018f57d18dc1bdd4ef35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72c4bd66488c4c14d6bb018f57d18dc1bdd4ef35"}]},"branch":"refs/heads/master"},"59a65b67dcb40881ef04b2e60a58ed7a0f43ee75":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-20 08:18:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/752955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/3"}}},"commit":{"parents":[{"commit":"81d6fa9a07c4f20d6bc244799d149101d0f62dd3","subject":"CI: Avoid molecule 3.5.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-20 08:18:24.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59a65b67dcb40881ef04b2e60a58ed7a0f43ee75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59a65b67dcb40881ef04b2e60a58ed7a0f43ee75"}]},"branch":"refs/heads/master","description":"Rebase"},"a14b2b6dae0893db0237cb2504581fca8692ce86":{"kind":"REWORK","_number":4,"created":"2021-09-27 12:02:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/752955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/4"}}},"commit":{"parents":[{"commit":"81d6fa9a07c4f20d6bc244799d149101d0f62dd3","subject":"CI: Avoid molecule 3.5.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-27 12:02:45.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a14b2b6dae0893db0237cb2504581fca8692ce86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a14b2b6dae0893db0237cb2504581fca8692ce86"}]},"branch":"refs/heads/master"},"ea563c54ac9eae56f8e7608c05474fe6fc28a533":{"kind":"REWORK","_number":5,"created":"2021-10-05 14:43:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/752955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/5"}}},"commit":{"parents":[{"commit":"81d6fa9a07c4f20d6bc244799d149101d0f62dd3","subject":"CI: Avoid molecule 3.5.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d6fa9a07c4f20d6bc244799d149101d0f62dd3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-05 14:43:52.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea563c54ac9eae56f8e7608c05474fe6fc28a533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea563c54ac9eae56f8e7608c05474fe6fc28a533"}]},"branch":"refs/heads/master"},"bc0885f5397af9ee69f8c873e57ad6f3d7a360e1":{"kind":"REWORK","_number":6,"created":"2021-10-05 14:45:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/752955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/752955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/752955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/752955/6"}}},"commit":{"parents":[{"commit":"3b901641025bc9dd2f9bcf1c53798958539bd3eb","subject":"Ubuntu: add upgrade jobs in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b901641025bc9dd2f9bcf1c53798958539bd3eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-05 14:45:00.000000000","tz":60},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc0885f5397af9ee69f8c873e57ad6f3d7a360e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc0885f5397af9ee69f8c873e57ad6f3d7a360e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
