)]}'
{"id":"openstack%2Fkayobe~753289","triplet_id":"openstack%2Fkayobe~master~I2c428ffc2b285aee03d8f59ae7cd3fb7230ce4ae","project":"openstack/kayobe","branch":"master","topic":"story/2008180","hashtags":[],"change_id":"I2c428ffc2b285aee03d8f59ae7cd3fb7230ce4ae","subject":"Add variables for API VIP address and FQDN","status":"MERGED","created":"2020-09-22 11:20:54.000000000","updated":"2021-01-21 20:16:29.000000000","submitted":"2021-01-21 20:14:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753289-story/2008180","meta_rev_id":"e9cf18448770adcc12574f3ce7415a3f063a3e4a","_number":753289,"virtual_id_number":753289,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 20:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-12-18 13:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-01-21 18:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-01-21 18:04:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-10-05 19:59:53.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 16:22:57.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-05 21:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 13:48:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-01-21 18:04:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b4a8fd32c3ea988be870cef5dbd5f83b8417837f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-22 11:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1be01c0ada2e64e4d1037ae89394de69f2ac835","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-22 13:02:19.000000000","message":"Topic changed from split-vip-fqdn to story/2008180","accounts_in_message":[],"_revision_number":1},{"id":"db579e5c8878d7a02a8a1de7212628822d3f8e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:23:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e88e5291914303a39e2ff31169a9ac : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5fa0f6f83d495c9611e7067fa83b6f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0347c049439f450b90bc785d21fc39a3 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5bec06b9db44a7b9a2f09d8926ebd10 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b598ac227441df81208e0556557ace : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b60c0e85df13417785cda24f51fb77fa : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/488a79625e4141f8932a7ba7500f4033 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5335f81896fc433499158845ce6399c0 : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9377661e5154cda845a8f0159082d0c : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/df6ae026d165455b809ec50598546873 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/986375a14a6549779da6902b9ef84789 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/19f0e4b6dac442bdb9a0ee76e457c0e8 : SUCCESS in 1h 50m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/763fe396e35044699fb4ed2112a2fd2f : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b38e598f357440895ae2b4860e79966 : SUCCESS in 27m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6bb8fc0e30b34e3db25d84c42783a0d3 : SUCCESS in 11m 30s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"72dbf1bf3651fd7516202d704e90cb6bad3cca22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-22 13:53:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"009469a416bda0cdcfa011d266692f3e10845594","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-29 15:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4723e1723ec466cab3735e37a9c05363b386499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 17:40:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49c0eeed49074be28293e5b5d587d103 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36f8611c99646a7921c1efe24f316b0 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fdf8bd7a45f4b2d9606f8349d0cc12c : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8fc7acda0b4d3e9421f4c664fa3ed0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7c41db80af482095ddf140a359c5d8 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69211f22fc57403ea50ce5f103991d9d : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d841c3ed18642a78c3e1fc3e9ddb4f9 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88206ef4e3a647a6ad02a53399a5529b : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afba42d1256d4779b17a9223cbc495f1 : SUCCESS in 22m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c4c90c5ac5fa4725a01fec50644d3fdf : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/21fdb212d4e04daaa83215a055ab128d : SUCCESS in 17m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/60b64e7e0b3a46afbec52aaccf1ef5dc : FAILURE in 1h 50m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8209f1f470344d7598ecbc711da8f34a : SUCCESS in 17m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9a81eac272d34a5e89b23bb57e0af830 : SUCCESS in 27m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e2df9d2a6f94455b8567093d28c4550c : SUCCESS in 10m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"ef7a00e1b7926fea89897a71f6a32465e3272ad3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-05 19:59:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c2f5587e2d96ed2c8e086bb3f1e808da02046414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 21:47:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7524cef2fb0749e6abd92635735305b7 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f63650df0574339b24bdca545765622 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b23c3043298c4e0b961b207b47a0557b : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19c6fe5ad70b4fd6aeaef313d9710915 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919c3eb6f233477a927df78a88a14352 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6541c2043b0c4327b468066cd34dc841 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f5dffa47af2c411f806203b8d1d48778 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3745f9602aa54120bcd19b17ece41f3d : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f28fdf698fb1492ab2091da1b529c9ce : SUCCESS in 21m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8aae04a887424609a1ec195eb0ffb3a5 : SUCCESS in 55m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5830706560c34860a1890bf871ca738c : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/488ec95254d84bd08a8bd61707a3f014 : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/11ee60700b0c4236b4cf1c80060c53f2 : SUCCESS in 1h 23m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5c51199cb3ce44c4a00eaf090ad5c360 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d2df13f6d10e40d6b51551f3098af321 : SUCCESS in 24m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a9f1a76f914a42b0be6fe0795910ea7f : SUCCESS in 12m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"78003d4b3ba50f9a83d58fe57ee509000284acc5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-12-18 13:48:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6b4b70109b6c2b0ddaecc20cb4500a3f94b750d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-21 18:04:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99d959e3d0853605dc201b046f636671ea9152d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:05:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b25aaca5587ba0eddd6147f8a5660260470330b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:14:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd2b6b18fff4aa49986948daa381bba : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b64aaa547e841348fbf1a038c02b8c6 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b00ab39b75164e288e64954756f9e3ac : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1379635ad5fb496695df788f4f9b8473 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/694c0550bf7f45abaf5975b9277218dd : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b6621ebe1d8432089994b9703ebdd5c : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe7687a3f54844139bf868f142a95e3f : SUCCESS in 7m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0fe25252441a4560b6608c1161a3c7de : SUCCESS in 19m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/97279b2e848a4d6bbb589703ec306eb1 : SUCCESS in 1h 21m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1e01533baa4b432faed1655685ed2ec0 : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a3a4064809284db5a4307999043ecc12 : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/df0dc534226146cd9be656069183cb54 : SUCCESS in 2h 04m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/507bd3f0a475446bb834b10d553830f3 : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bb64fb30c0d048709cb182c42007266c : SUCCESS in 21m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cf971bb762c4403e8bb62dd16dcde714 : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"d53b6301e90a6a665c316b878ecef958d26ed031","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:14:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e9cf18448770adcc12574f3ce7415a3f063a3e4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 20:16:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2221abef07124a6cab56f4462a0e8eba : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f579c694693942ebbca60e783426b9ba : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1862e24bb5062be181e3462a773b96b214cd8be8","revisions":{"2d7ad7d1f5e80cd6817ca375c63cc5cd95f9fa41":{"kind":"REWORK","_number":1,"created":"2020-09-22 11:20:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/89/753289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/753289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/753289/1"}}},"commit":{"parents":[{"commit":"4390ad247a885843b9c143652d3efea64d2f0c0a","subject":"Merge \"Performance: skip LVM configuration if no groups configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4390ad247a885843b9c143652d3efea64d2f0c0a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-22 10:39:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-22 11:19:45.000000000","tz":60},"subject":"Add variables for API VIP address and FQDN","message":"Add variables for API VIP address and FQDN\n\nKayobe currently supports definition of various different networks -\npublic, internal, tunnel, etc. These typically map to a VLAN or flat\nnetwork, with an IP subnet. When a cloud exceeds the size of a single\nVLAN/subnet, this approach no longer works.\n\nOne way to resolve this is to have multiple subnets that map to a single\nlogical network, and provide routing between them. This is a similar\nconcept to neutron\u0027s routed networks, but for the control plane.\n\nAn issue arising from this is that if different hosts can have different\nnetwork definitions for the internal and public networks, it is no\nlonger trivial to use a network attribute [1] to specify the VIP address\nand FQDN. Furthermore, the play that generates Kolla Ansible\u0027s\nglobals.yml containing the VIP and FQDN variables runs as localhost,\nwhich does not necessarily have the internal and public networks\ndefined.\n\nTo resolve this, we add global variables for the VIPs and FQDNs. The\ndefault values are as before, except in the case where HAProxy is\ndisabled, which we no longer provide a useful default for. That\nconfiguration is very rarely used in practice, and the need to reference\nthe IP address of a host in the network group makes it difficult to\ndefine safely.\n\n[1] https://docs.openstack.org/kayobe/latest/configuration/reference/network.html#global-network-configuration\n\nStory: 2008180\nTask: 40937\n\nChange-Id: I2c428ffc2b285aee03d8f59ae7cd3fb7230ce4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d7ad7d1f5e80cd6817ca375c63cc5cd95f9fa41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d7ad7d1f5e80cd6817ca375c63cc5cd95f9fa41"}]},"branch":"refs/heads/master"},"def5965f2017902b3f591a5d874ade8b5956c85d":{"kind":"REWORK","_number":2,"created":"2020-09-29 15:43:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/89/753289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/753289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/753289/2"}}},"commit":{"parents":[{"commit":"7c25fec6ca9526e7c7f12c05d45feb6063a88974","subject":"Merge \"Remove unneeded doc requirements in molecule tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c25fec6ca9526e7c7f12c05d45feb6063a88974"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-22 10:39:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 11:06:35.000000000","tz":60},"subject":"Add variables for API VIP address and FQDN","message":"Add variables for API VIP address and FQDN\n\nKayobe currently supports definition of various different networks -\npublic, internal, tunnel, etc. These typically map to a VLAN or flat\nnetwork, with an IP subnet. When a cloud exceeds the size of a single\nVLAN/subnet, this approach no longer works.\n\nOne way to resolve this is to have multiple subnets that map to a single\nlogical network, and provide routing between them. This is a similar\nconcept to neutron\u0027s routed networks, but for the control plane.\n\nAn issue arising from this is that if different hosts can have different\nnetwork definitions for the internal and public networks, it is no\nlonger trivial to use a network attribute [1] to specify the VIP address\nand FQDN. Furthermore, the play that generates Kolla Ansible\u0027s\nglobals.yml containing the VIP and FQDN variables runs as localhost,\nwhich does not necessarily have the internal and public networks\ndefined.\n\nTo resolve this, we add global variables for the VIPs and FQDNs. The\ndefault values are as before, except in the case where HAProxy is\ndisabled, which we no longer provide a useful default for. That\nconfiguration is very rarely used in practice, and the need to reference\nthe IP address of a host in the network group makes it difficult to\ndefine safely.\n\n[1] https://docs.openstack.org/kayobe/latest/configuration/reference/network.html#global-network-configuration\n\nStory: 2008180\nTask: 40937\n\nChange-Id: I2c428ffc2b285aee03d8f59ae7cd3fb7230ce4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/def5965f2017902b3f591a5d874ade8b5956c85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/def5965f2017902b3f591a5d874ade8b5956c85d"}]},"branch":"refs/heads/master"},"1862e24bb5062be181e3462a773b96b214cd8be8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-05 19:59:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/753289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/753289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/753289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/753289/3"}}},"commit":{"parents":[{"commit":"1f155b5582668eddc72f5bdd63946632288a2dc8","subject":"Merge \"When EPEL install flag is not set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f155b5582668eddc72f5bdd63946632288a2dc8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-22 10:39:31.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-05 19:59:53.000000000","tz":0},"subject":"Add variables for API VIP address and FQDN","message":"Add variables for API VIP address and FQDN\n\nKayobe currently supports definition of various different networks -\npublic, internal, tunnel, etc. These typically map to a VLAN or flat\nnetwork, with an IP subnet. When a cloud exceeds the size of a single\nVLAN/subnet, this approach no longer works.\n\nOne way to resolve this is to have multiple subnets that map to a single\nlogical network, and provide routing between them. This is a similar\nconcept to neutron\u0027s routed networks, but for the control plane.\n\nAn issue arising from this is that if different hosts can have different\nnetwork definitions for the internal and public networks, it is no\nlonger trivial to use a network attribute [1] to specify the VIP address\nand FQDN. Furthermore, the play that generates Kolla Ansible\u0027s\nglobals.yml containing the VIP and FQDN variables runs as localhost,\nwhich does not necessarily have the internal and public networks\ndefined.\n\nTo resolve this, we add global variables for the VIPs and FQDNs. The\ndefault values are as before, except in the case where HAProxy is\ndisabled, which we no longer provide a useful default for. That\nconfiguration is very rarely used in practice, and the need to reference\nthe IP address of a host in the network group makes it difficult to\ndefine safely.\n\n[1] https://docs.openstack.org/kayobe/latest/configuration/reference/network.html#global-network-configuration\n\nStory: 2008180\nTask: 40937\n\nChange-Id: I2c428ffc2b285aee03d8f59ae7cd3fb7230ce4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1862e24bb5062be181e3462a773b96b214cd8be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1862e24bb5062be181e3462a773b96b214cd8be8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
