)]}'
{"id":"openstack%2Fkayobe~754053","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Iae75e17024adee9c2874c14d3ed36f4c87ba48d7","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Iae75e17024adee9c2874c14d3ed36f4c87ba48d7","subject":"Performance: Parallelise Kolla Ansible host vars generation","status":"MERGED","created":"2020-09-24 09:54:44.000000000","updated":"2020-10-01 11:41:49.000000000","submitted":"2020-10-01 11:40:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754053-1601552424577-48b48ccc","meta_rev_id":"dea0d65939616fad674488effd2eb36e3b25e1a0","_number":754053,"virtual_id_number":754053,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-09-24 16:21:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 11:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-09-29 10:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-01 09:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-01 09:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 10:16:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-01 09:53:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-01 11:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b49f74d40553921ea859a202a7cd7b89c44dfbcd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 09:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32cc88c0faf08de471d0f4f7c156088d4eb60c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:54:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cd87b8880cb4ae48bec2de8c61839b5 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7839399277e247acb57cdf3d414829d8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b89a8888654da6b40707f1c7ff694a : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34a3ac30a1934f718ca21406af2c377f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86c91d4d4e2447d89d3413ffba353bc : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a1938a2df774e00baaa4d05bf7feb1c : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f643efa7006f4f7b80adef2986c82b1f : SUCCESS in 7m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5c845987a864d8cb4176b33075dee3b : FAILURE in 8m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9096d0c69cc84fe28ab21ff6e30c45d9 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a94ee9e506094fc1ac955636290f338b : SUCCESS in 19m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2813daf69135459d8acd9f3d533da0d3 : SUCCESS in 1h 50m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8598035ffd4f4a25bef82fec39839c02 : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d25fdc227c1541aa899a954f8affcfe8 : SUCCESS in 29m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/44c9035e11ef4ee48b778b321f763932 : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"44ebae459e6d32e2a50e3c2c25b1d4d072812591","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 16:21:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dce52a5b20e9c8fa58da0c2768314eceb30ec61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:17:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be62ec8e30342bca27a0614002e8583 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8bca81042c401ebfaea0a4f8058736 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6138e265e24b76bb30fcf7a9a88e79 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/995d20fde7734d0184f02bd8ee38e97c : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c12dc45a97417cac8aa17d185b2538 : SUCCESS in 8m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab86a4cf5718483d92810029575f46dc : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/839bebe4f51242799c4659ab1916eee3 : SUCCESS in 8m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdb6a29f9e3d42b7a3b5440dcf168eea : SUCCESS in 23m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/007180aa93b046c194a1800b882d940d : SUCCESS in 59m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7c130372b4dd4705b3e2ff03ff40e14a : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/572c96f106e94c119fd742f46c8657c5 : SUCCESS in 1h 40m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ccc31020b0934ba7b600c9c0a5b7a572 : SUCCESS in 17m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ff2e4a89760e4497a23511665118e1a1 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ae3478b9b5a2422aadcd2b3d39b0a761 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e92e7a9ae359642e926829534892fedc26bcc3f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-29 10:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e807a66bdf8ae1527caae5a52d6afe4b8d6c411","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-01 09:53:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"847f989941ecbbe9de24b2ec13764e6e30c22a4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 09:53:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c71fc24a6e6565ab99233228bc40745c3bb4b4ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:40:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9eaba04a8af47da919652454ccd568c : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2361c81fafe4ca590174b4fdb3f62e0 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46f9b4e314a9456cbef5a56d35a0703d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553000da1d014c85a5a83b15efc0d9e2 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/074542204154464f9d28e52f97a09896 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c7d4980f78f45a3b5c61e4acc88b974 : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93435252e3814520976dfd5ca3ffc8b1 : SUCCESS in 20m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/78a3a3083bef4d4bb0dd4bda068591c0 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ffc02f2769a7409b86075667b581824f : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fed56703da5d478da78810fa544a42e3 : SUCCESS in 1h 45m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/536c1bf8755b40bea6106f1f3635e192 : SUCCESS in 16m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bd6d14abc5f44d8cb3c750d452aa69f9 : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/37d1c4ba76474211a5651c482d5bf8eb : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9280cc141cf41b7a39b925a61a380ca37f370c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:40:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dea0d65939616fad674488effd2eb36e3b25e1a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:41:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d8cfa9e825426094705f0e86b41437 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41c889241f4f636bef5b10955d317b3153089dc5","revisions":{"41c889241f4f636bef5b10955d317b3153089dc5":{"kind":"REWORK","_number":1,"created":"2020-09-24 09:54:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/754053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/754053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/754053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/754053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/754053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/754053/1"}}},"commit":{"parents":[{"commit":"d40736b75555bafdc55b0321fe2551c870d0ebc8","subject":"Document that kolla_tag may need to be changed when upgrading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d40736b75555bafdc55b0321fe2551c870d0ebc8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:03:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-24 09:54:22.000000000","tz":60},"subject":"Performance: Parallelise Kolla Ansible host vars generation","message":"Performance: Parallelise Kolla Ansible host vars generation\n\nKayobe generates a host_vars file for each host in the Kolla Ansible\ninventory. These contain network interfaces and other host-specific\nthings. Currently this is done by iterating over all hosts, which does\nnot scale well with a large number of hosts.\n\nThis change extracts the host vars generation into a separate role, and\nexecutes it in a play targeted at all hosts, with delegate_to:\nlocalhost. This ensures that host variable files are generated in\nparallel.\n\nStory: 2007993\nTask: 40629\n\nChange-Id: Iae75e17024adee9c2874c14d3ed36f4c87ba48d7\n(cherry picked from commit b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c889241f4f636bef5b10955d317b3153089dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c889241f4f636bef5b10955d317b3153089dc5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
