)]}'
{"id":"openstack%2Fkayobe~754064","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Iae75e17024adee9c2874c14d3ed36f4c87ba48d7","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Iae75e17024adee9c2874c14d3ed36f4c87ba48d7","subject":"Performance: Parallelise Kolla Ansible host vars generation","status":"MERGED","created":"2020-09-24 10:40:42.000000000","updated":"2020-10-02 17:50:42.000000000","submitted":"2020-10-02 17:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754064-1601660951610-8f1aa1ae","meta_rev_id":"ccc442b90d0955f5ec7dc79a9eae46125bada355","_number":754064,"virtual_id_number":754064,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-10-02 14:25:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 17:49:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-09-29 10:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-01 09:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-01 09:53:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 10:15:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-10-01 09:53:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-02 17:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56aca4606e7c714cf0c926007cb2ba76fd7b4342","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 10:40:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d95ac42b6841a92e0febc36a2e76b5f500648a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9ab95edd0246f8bd558e642a9f50f5 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd449ad47ccf4e35b0d226fb2c5ba607 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3ce0d6aea7489789d41a147ea62c46 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de4e06633524cdf9718295978a07b4f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60cc25e8026b42638765e0743c14b70a : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eedd54c376424252be1133788f046ebc : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/002b3104f8124287a706607cedf7b966 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d047215b7b844072860ab053eb62842c : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8de5790d9e0445a0965b7d69c19f0681 : FAILURE in 7m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7d959d0fb425456a8abd9d19512a9155 : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3643a074e71640529695df32a11f4827 : FAILURE in 22m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/01b10e504e734e65ab93692dbeb86ad0 : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed24872255f940588ed78074acfab2f5 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0270e81d13e5413dbb70634301f3f38b : SUCCESS in 1h 39m 00s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2fae46229e4d49cfa22776e70db9c070 : SUCCESS in 23m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a3eca7a7f04f48069c30280f45d3a42a : SUCCESS in 16m 28s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d0cd7583d46244ce9e8defd60638eb53 : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cb911fdafadce44820f959d3589d59adff472123","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 16:21:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3325801a733a742b51c936f02d664df7c904d50b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:14:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/708b6e0cd5b04d1eb6eea13581a198f0 : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2559b5a14844418934b40276e3698a6 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aca5bef0d184f0682df2f717a949e0e : TIMED_OUT in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96dc342519fe4c52984de36ea0d1a7db : TIMED_OUT in 31m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51bdd9c51b2c41c4ae5252131bb833e2 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f55290a4899469eaeab44d1204ec857 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/427726e9f63540c6a59aa45b74158d52 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0fce0640e9747968c89b354becfd6d5 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6d5672c5bec4619880f82f69280b2e6 : SUCCESS in 19m 32s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/086071fcca24454aa095dfdbd0e4a42c : SUCCESS in 1h 00m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0f993506bb8c461bbd639e4c1aae06ca : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/0ad407843cc24f30aa3de11238ab76b6 : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3eed384a70514d0c9f6f8fef3e43fc48 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a3d82c04d89148dfbac6232f4d23d51f : SUCCESS in 1h 36m 03s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c1b807a1894b460496e68fad01a4e1bd : SUCCESS in 19m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/27bdb647a87e4cdb8e0ef8a45b30d146 : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/149b7e0c587c4d76a45f5d94c3cb8d3f : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"356f4bc64c29b76ab3202a6b64afc2eb9e8e9402","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 18:17:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f601f528d8cef151f691f88abbcfb23850c819c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 20:04:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9ee38c74c7f4ddfa8529dbd1c5ab70d : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/00917a0250b04dac89bd7766d2a48d2d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a87f0d5c015b4ecd8729545e87c99524 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd36940bcdc4a45a92fd7999b95a0f1 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12011a4998a542e3a5e3e1ff24c3eed9 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995c5b05cdf642da9bfd565311e7dba4 : SUCCESS in 9m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3da26fa796814756b7db38b8e46f50db : SUCCESS in 8m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9a56c82d2f341d78671bca1ddd56e39 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cffd57a195aa4a4499b358b7974761b7 : SUCCESS in 20m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e1b7c4eec19643f3bb24a4afbf252e08 : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/52686e7e45924a999ab0a8905eafd0e9 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/b0ce4567fdd84fda9f5bf290431e1d1a : SUCCESS in 19m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fb3e94c0e43b4c689b470c7db65f18bf : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/28ff1c7d1ad14eb89af5755b608c785e : SUCCESS in 1h 35m 32s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ee594b6cee2444bc84d4c1f4c9051acd : SUCCESS in 15m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0d90b02d4d444811861108f5df42ed5f : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1289613651584c00be44d0540297f39b : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"de538dd06a7faca59d452f081c5c6c8a0ee2f242","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-09-29 10:15:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8afa5a9411ec8baca45a64104b10ffd661051e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-01 09:53:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7722dc9340eb8848fcb9aa31b984d07c3c4db7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 09:53:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4324e8669235830c4294610a28495e98637a6da9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:51:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c7c5476b277458fb94d113b6df17775 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/416e3e6e450042db99f7c3924b0c6abf : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd9de49ba6b4da8b8cbf157c113ed36 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f7cb2ad2168410c83ee1af45bde59dd : FAILURE in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fadd27260d4ce5b22ea6604e3d888d : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b96fcbc8c2da4937a5a40ead02c21348 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/401482ee3e0e48878c774e42176b4dc9 : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/847936c18e014d288f2c92203c599b97 : SUCCESS in 18m 26s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f20fa208eb04430d8ee25d0ce1f21bcf : SUCCESS in 1h 13m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/35120e3089614a5281c3b1b16cb5e440 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c2bf403fbf8f4a63847c2832b2c5fe5e : SUCCESS in 14m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a48b53bbaa224c94968232078268eb8c : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2f7c9d69acb74040a688dc7b64fc96b9 : SUCCESS in 1h 52m 35s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/e46cd4b9f1bc4f57a57df62d7a3b4c79 : SUCCESS in 23m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5bfdc6620bd44ec490f6a6c1ab857d07 : SUCCESS in 19m 39s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/15396dcc75c544e99d5fc4cf686687b1 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c368938d281906247c23f87cd1118cc219e467fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 14:13:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06973f8aebd80e02ccac059070ef810e6f5b9660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:12:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5de66da6f1b4a7096f14a6ca29b2d4d : SUCCESS in 3m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd06b94869054b7a98d9fe390b65440c : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e8edffac81411287abfba4548f04d4 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba08d1a6a8040888203488b4fc9824b : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e082e9df9b0447308393b7a0c6a8efb4 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85abcbada7684abaa67aa993f67718d7 : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a840ac2651784307953f8801a5685320 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69703adfe0724130aec65a929f74d227 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a81153ecd364816b1b0f599034e7dd3 : SUCCESS in 20m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/39e6f90452204c5d9c3569464ddacd19 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/423ff651b46f4f568f5ad500a331878b : SUCCESS in 1h 08m 58s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/12ae610570334f8b8427dc8cc41b380f : SUCCESS in 15m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1e6ba468868b461d9e5e9893f4071f56 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/69d21e2b277b41d2b17e9e2141b485a5 : SUCCESS in 1h 52m 20s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/175dd84fd34742869de09261e62800d5 : SUCCESS in 19m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2375c93987114eee946e6dc8fa3dd92f : SUCCESS in 16m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/48213ce5d5ad4e569b9732389e015a19 : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":1},{"id":"638d530b23c0163f51d5b885c6d65f5983ed0b86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:13:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0c72342c648e8f8e096ff5f68f699d8bb14bfbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:18:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2aedbcf32fb4d69b69d2d839f8f7b10 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/183597dc615f46099fa5a70c4e49e2b5 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eaf05e6dde34425ad7e41d6a0b093f6 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8e4cca98d45467595ab4dbc7f02c88c : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e0756604604302b0318e5a208e287a : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb0b56b6460f4f18ba9aff1099e8dcb0 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c383b33ded10496e898ae6548c8e7025 : SUCCESS in 9m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aac356f031dc4fa1a4457104c1bcb340 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7a15e21aa58e478ba41613600a4dc7ff : RETRY_LIMIT in 17m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a7d74ddd78484fa9aca42fb284f26cbe : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/7450ae832cd041b8aace1a8f7ecefe87 : RETRY_LIMIT in 15m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/11cf86fa9d2946f0ac2314420d5820df : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bbd54b38ce0e4ecd916228f212d92b9b : RETRY_LIMIT in 15m 33s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ecc388565f57486caeed3877ede1e573 : SUCCESS in 25m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/abeeac76e9f040a7938ee8f5dbe40f6e : RETRY_LIMIT in 5m 35s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0ddeccb37f7446f8905bc2fe9930fa7f : RETRY_LIMIT in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"de1ffb42f605335d6c45f2d2260e409fd050f35d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:25:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d256d72419702b1decd384160af4c8796d466ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:19:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5da39fe17844329ae1fc694d9b7b985 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08197dbc785940bea29460a4e51c0f6c : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f5c4da44ad14ac9ad3467fcf35a7a7d : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251a54129da94d9f9316112ad5ebc928 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/926fdcc9d0ad4a26954a4c2761cd3393 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c5665d6c38439ca4e1a9e810bab278 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0f2175dd4b34f4dbf751c6ff18cc63f : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c22e79e34df2429ea3ba6b49d8c8372b : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d4a1572d3e24359a549dc01b7c43638 : SUCCESS in 17m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/334c8a8fcbed4e7aa232e479b3ec5bfc : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c5218a02b9984299b161e64b334ea2c3 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/115d7b1f95aa4cb1930835d5ab523dd5 : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ba221b2429934398b996aa85b1e50c13 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/80ca7deafbe24a15807a40cc6d7ea3c3 : SUCCESS in 1h 43m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/cb39da4229744617a08ff19a945f1c1d : SUCCESS in 21m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d9b60d193ab5407491f139faa6d31bc7 : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3eda8d6932844d61800c87b1d4b32291 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0c6021fe9a804ea3c022fee6dcbdb17a888fb094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:19:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff3b83ff5adcf94863c904a44c096c7e6958c8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:49:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/acf41b683e3541f989fadc56edc51095 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfb0905d477495ab75b7fbaeb60e029 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272dbd740573451eb79daecd0c872b0f : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c7d0ace57f424e85abebd0c0808fde : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde66f4286dc4fb5bb0066e95da8d104 : SUCCESS in 18m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6aaeeabe7abe46e2881311da921b7ebe : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7929bf988ad4412d90323ac134e3f9dc : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/373f5e071bde4271b2216cd34b841607 : SUCCESS in 17m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/324c69382b604025afff59893a078b7e : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/10204184c15647b688dc2d9fc53f8132 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/cb20b3801b034c1cbc401d683c6af9f9 : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/076e32818a9b45a290ccda53a196bba6 : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e0994d2998f84b38bec74850aca746c3 : SUCCESS in 1h 26m 31s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3dff2e9fe76c4f2389761fcab07cc321 : SUCCESS in 22m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7c7aee8af6eb487a80106d628d89241d : SUCCESS in 16m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/959c356690da4e93a831792a0d057a5d : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2ae5681f344762dd0a3fcd173a934729a7e4f7c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ccc442b90d0955f5ec7dc79a9eae46125bada355","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b00af48e9314e0d90139313ff032e96 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74f56030827a82ecb48c41176926a8be666db14f","revisions":{"74f56030827a82ecb48c41176926a8be666db14f":{"kind":"REWORK","_number":1,"created":"2020-09-24 10:40:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/754064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/754064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/754064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/754064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/754064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/754064/1"}}},"commit":{"parents":[{"commit":"94848eefecb0d757f908d0f21c5205cc07e05269","subject":"Document that kolla_tag may need to be changed when upgrading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94848eefecb0d757f908d0f21c5205cc07e05269"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-08-04 16:03:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-24 10:40:30.000000000","tz":60},"subject":"Performance: Parallelise Kolla Ansible host vars generation","message":"Performance: Parallelise Kolla Ansible host vars generation\n\nKayobe generates a host_vars file for each host in the Kolla Ansible\ninventory. These contain network interfaces and other host-specific\nthings. Currently this is done by iterating over all hosts, which does\nnot scale well with a large number of hosts.\n\nThis change extracts the host vars generation into a separate role, and\nexecutes it in a play targeted at all hosts, with delegate_to:\nlocalhost. This ensures that host variable files are generated in\nparallel.\n\nStory: 2007993\nTask: 40629\n\nChange-Id: Iae75e17024adee9c2874c14d3ed36f4c87ba48d7\n(cherry picked from commit b12f9e35e5d6cf49e72c9ccf4988365c850b8e1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f56030827a82ecb48c41176926a8be666db14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f56030827a82ecb48c41176926a8be666db14f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
