)]}'
{"id":"openstack%2Fkayobe~754082","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53","subject":"CI: Fix kayobe-tox-molecule job","status":"MERGED","created":"2020-09-24 12:50:23.000000000","updated":"2020-09-24 13:44:38.000000000","submitted":"2020-09-24 13:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754082-1600954954330-e805a381","meta_rev_id":"0a88c9556db8e329ba5c0b4087ffdfe73f9938c9","_number":754082,"virtual_id_number":754082,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 13:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-24 12:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-24 12:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-24 12:56:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 12:53:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-09-24 13:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fae5e7583cc9249dc0985c5ff181b3a8e2e34447","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 12:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0df82c93672a53461c36bfff3679392614342425","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:51:55.000000000","message":"Patch Set 1: Code-Review-1\n\nCommit message needs updating","accounts_in_message":[],"_revision_number":1},{"id":"0b0017135702ef512d7645d7075ec29c3c535be1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:52:55.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"be0907cecfb07c55b85a6fb3b013d1ccb121bbcb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-09-24 12:53:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20ff66384d39df0d51edd367e39c2788866e9930","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-24 12:56:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15bceebf4359e3c4f5df09d47dfdda0f9153f8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:16:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0623ee082af44ce5bac45d0ef5d15f4c : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c109dffc0e4946f18f38c553445e9508 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5820118c4d8f45ab83952221f653f972 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b788f5dbca624ff79417e82607e07d41 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b048a6c0b7a40eebd5b07c4e021d026 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112f1e333f7645fd8b8f1825a2cd7770 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1b3b98916242868558d1efe8a08ddb : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9dae1d1c92fb4df3a543d12c1a703ee7 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/901cf8833eb347ea8ba86918ee713273 : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2527a899bf2747dc9570c5b6ae04aa39 : SUCCESS in 21m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8f86a5307dc3d27255a45f40ae3a77b2238ca401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:16:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53088142e5e296a8d0b8d9c40ad459268a305779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:42:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/437f83fea67f442f90e30417c65102fb : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18103b575cf1492199a47d6fbcc308b5 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87532ae66394816bdad78646f35f016 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6e7d62e1ac64413b9552ad62ede0aa6 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f3d81023fa49598e6801fdc0fa9d8f : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3637ce5f60c483392801dfb85efd2e6 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98773d6e81bc47ca970b24588bc94b33 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f614c1464c5746e3b63854fc7cc1e282 : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb9818bcfb524dc1b6eb48ec4cc5fcfe : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d9b78b6313e4b693198bb3762cd64df9223a9218","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:42:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0a88c9556db8e329ba5c0b4087ffdfe73f9938c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:44:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c35939da9284cbc95f305b2d2e7c1f7 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80d57d321f7c43148c88f1248293e7b4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b209dad81c0af644f2b3e0bdae07765f55ff4ff","revisions":{"fef0acd5f8de75c9eea0f1dfe681a9ea533b5788":{"kind":"REWORK","_number":1,"created":"2020-09-24 12:50:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/754082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/754082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/754082/1"}}},"commit":{"parents":[{"commit":"94848eefecb0d757f908d0f21c5205cc07e05269","subject":"Document that kolla_tag may need to be changed when upgrading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94848eefecb0d757f908d0f21c5205cc07e05269"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-23 16:34:23.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-24 12:50:15.000000000","tz":60},"subject":"CI: Fix kayobe-tox-molecule job","message":"CI: Fix kayobe-tox-molecule job\n\nWhen molecule runs, it installs ansible\u003e2.8, which is version 2.10.0\nsince today. This now also installs the ansible-base package, currently\nat version 2.10.1. Then it installs the version of ansible required by\nkayobe, i.e. ansible\u003e\u003d2.8.0,\u003c2.10.0 (currently 2.9.13).\n\nWe end up with a tox molecule venv containing:\n\nansible\u003d\u003d2.9.13\nansible-base\u003d\u003d2.10.1\n\nThis causes the following issue:\n\nERROR! Unexpected Exception, this is probably a bug: cannot import name \u0027AnsibleCollectionLoader\u0027 from \u0027ansible.utils.collection_loader\u0027 (/home/zuul/src/opendev.org/openstack/kayobe/.tox/molecule/lib/python3.8/site-packages/ansible/utils/collection_loader/__init__.py)\n\nThis commit adds requirements.txt to the molecule tox deps, which\nensures the correct version of ansible gets installed first.\n\nChange-Id: I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53\n(cherry picked from commit c58555498836066002dc33620018e9bd1e2eeb6a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fef0acd5f8de75c9eea0f1dfe681a9ea533b5788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fef0acd5f8de75c9eea0f1dfe681a9ea533b5788"}]},"branch":"refs/heads/stable/train"},"0b209dad81c0af644f2b3e0bdae07765f55ff4ff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-24 12:52:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/754082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/754082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/754082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/754082/2"}}},"commit":{"parents":[{"commit":"94848eefecb0d757f908d0f21c5205cc07e05269","subject":"Document that kolla_tag may need to be changed when upgrading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94848eefecb0d757f908d0f21c5205cc07e05269"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-23 16:34:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-09-24 12:52:53.000000000","tz":0},"subject":"CI: Fix kayobe-tox-molecule job","message":"CI: Fix kayobe-tox-molecule job\n\nWhen molecule runs, it installs ansible\u003e2.8, which is version 2.10.0\nsince today. This now also installs the ansible-base package, currently\nat version 2.10.1. Then it installs the version of ansible required by\nkayobe, i.e. ansible\u003e\u003d2.6.0,\u003c2.9.0 (currently 2.8.15).\n\nWe end up with a tox molecule venv containing:\n\nansible\u003d\u003d2.8.15\nansible-base\u003d\u003d2.10.1\n\nThis causes the following issue:\n\nERROR! Unexpected Exception, this is probably a bug: cannot import name \u0027AnsibleCollectionLoader\u0027 from \u0027ansible.utils.collection_loader\u0027 (/home/zuul/src/opendev.org/openstack/kayobe/.tox/molecule/lib/python3.8/site-packages/ansible/utils/collection_loader/__init__.py)\n\nThis commit adds requirements.txt to the molecule tox deps, which\nensures the correct version of ansible gets installed first.\n\nChange-Id: I7ef890a2ae6fd67f53e1b8bca244a1c0d44fea53\n(cherry picked from commit c58555498836066002dc33620018e9bd1e2eeb6a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b209dad81c0af644f2b3e0bdae07765f55ff4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b209dad81c0af644f2b3e0bdae07765f55ff4ff"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
