)]}'
{"id":"openstack%2Fkayobe~755015","triplet_id":"openstack%2Fkayobe~master~I300d0ef136224126f25d2c70a80a42afeea5f586","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I300d0ef136224126f25d2c70a80a42afeea5f586","subject":"Disable overcloud networks for Ironic when unused","status":"MERGED","created":"2020-09-29 14:01:21.000000000","updated":"2020-10-06 14:23:06.000000000","submitted":"2020-10-06 14:19:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755015-1601993941851-e81968aa","meta_rev_id":"ef4d34e7b51ca3d58b545f1418f0623a9b215885","_number":755015,"virtual_id_number":755015,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 14:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-10-06 12:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-10-06 11:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 19:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-10-06 12:18:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 19:57:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-06 11:24:57.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-10-06 12:18:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-06 14:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63136570afde87d199309a318efd86a3b2e0eff1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-29 14:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1b9ffd49872a68adac3a403c3f303b2d3027413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 15:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e28d54586408461f9d217c78ea7e2d57 : FAILURE in 38m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db5ec76d5ed24f6b816760bd4fb372ea : FAILURE in 47m 53s","accounts_in_message":[],"_revision_number":1},{"id":"36a3a6839d59dd1d9aaddf14639f2327453b350a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-29 16:22:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45b82674e5a2461b3701f005c01157a7ac2b5241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 17:49:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f8ece0c73a24392b0ae401456a5eb9e : FAILURE in 42m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/22e65a9ec9fa44578171e20b90f0d2bd : FAILURE in 47m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5a81c13e9dc2f05159f4c4f24c4901783319ba7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-29 18:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a04115628262b97b5241dabd205340ba3746971f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 20:05:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b573ae4d48b84ad48404908858e15b82 : SUCCESS in 47m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46abc94be8ff42ab81263bd038af11b2 : FAILURE in 46m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7e79d873824735e4bb9ebe8a11c4f1c232676efc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-30 08:09:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"faa6035ac07efd39dcb755da540208e371413dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:37:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/73a94eca64e84e75934adfadfa8a1797 : SUCCESS in 57m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/88ab6167894c445ab51e597cfa0aae99 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":4},{"id":"84900828b5b19d7b21b9230eb5729ea9c07a3e59","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 10:17:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e2aa27818b95ec06d65b264b2e9d7d7e88c0e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:45:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739eb68a409840df90c3cbf207129d5d : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b833a015552748f784a18db413b086a4 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380a4b5960fd4e35bd31e44a7bffb93c : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56ab22a0d5c94ead8747af647f41689b : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07438f3fe144d40b184ef11cc7ae89a : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/183aad27cb4842ce89df973060c6152b : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e51af9981fc84ff6af0881056a6b7bb6 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c866205c4aa44878943e4275a965b75 : SUCCESS in 6m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/121c3697b0af45e1bb23efb370a255e3 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7fff781bc02d40f38359537d024759db : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/50fd2ccc5ccb451ea6b1c77f41191eb5 : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a31f36c7cf6c4906993abd93d665a0f4 : SUCCESS in 2h 08m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ac678b62f41c4f52bd54a145525ec081 : SUCCESS in 17m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/44338addf54c45f9ad6dbe8bfd74b452 : SUCCESS in 29m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3c791c8098c54d53a198fb686574ebc4 : SUCCESS in 9m 58s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"96f9c2f66dff4f6e1e3dc712a4e62d180118949f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 10:03:18.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"88e05ca6af1b72914a871b76ebd3feb460880c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:31:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80d466ab1f7d4692bb309d005f551f43 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97aed1e2320c4bee9b6ca3d28c4aa6cf : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fde3cdd334f42f892f7e543806fb422 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ba58ae36e6d4c76bf4034eaa9786f58 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565a03cea8ae4f6f989b68337b241d92 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/077816a34afc4b2fb5e773e69d4fac45 : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75b15a642c1c48f8b3a19006e81f9d41 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1556d8163ec4fbba82d3b38ccb5224b : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f9b4fd424274c428bff37c2613fd31a : SUCCESS in 20m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e4f308d89a414327b429bc46f77d87c3 : SUCCESS in 58m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e609b065ea234ac094a9564696b6030f : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/253cd27836c34d2994a8287ee9b141b6 : SUCCESS in 14m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/26cd7938c94d4953b8de637ed08ee238 : SUCCESS in 1h 17m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cd7ac17d33014d8ca754452be9e26831 : SUCCESS in 12m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cedeccf6adf541ad9517b156e35c4ff5 : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/217b3ee2580b42188f627ee11d4d52c5 : SUCCESS in 10m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"9594bf076c4143086ae0e9fb838efa223e6cf186","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 13:41:29.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nSmall comment about release note","accounts_in_message":[],"_revision_number":6},{"id":"5556e74959667581e8c85624ac4333b0d783dba1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:22:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b81f654c1a084604d85b21da06b5643fb647f924","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:23:14.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"266013ab3ad81a0e0733b102e913fc29103be779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:22:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d88ce4018540c48a2f01e21b19e604 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06699efeeaa444cbc49e007a07f5c68 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dfafb1617d34b3eb17727a7b535ab20 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5901be9957fb4fc1a719166c2ae5e435 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038c7e5b653049b58f98f719161ce69b : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b0dc9d2205a484ab37071d46a6b0f18 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/127b6243f76d411a94d1088a56e306b5 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72c2f63b77d94a149402206224adf10b : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a92a40c827e64f959eb6430c54177f45 : SUCCESS in 22m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/48549588a80246f6a675e202689be01f : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cddad0c16cf1494fac1f6075ab7fd2f9 : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a7666b6a3e0f4ab4837b6245ad6f6895 : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/742910e69de240be96ddcc8f95dc74aa : SUCCESS in 1h 52m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/84302c0f6cf64b27a3e466a835f5e4bb : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/16cab86a5b9b42859f528a4034af0197 : SUCCESS in 27m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7adb2053e6e5465cb194320119875c9f : SUCCESS in 12m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"631e1c632f6f0252601ce7384d6c84aed38959e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-05 19:57:49.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"cffe397f950c9ef88107de484db882c638415d5c","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-10-06 11:24:57.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"3fce89c38fdea42ca711f77b774a57f0eac06c32","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-06 12:18:20.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d04dedc8372aba94e9f74669d02a9d6e76c567fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:18:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ff3b497fe536bd71e1b43b234fe79dc3abd66239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:19:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9e043a0fb3ff1ad9e5f6071d52cf016260014479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:19:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d89507725654e129f8eca273ded4161 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b73bc5e7e2894196a82d80b0e3595fd9 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6b932eb95245df99169ea0f166769d : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c977b1a973441e8d149efe38942328 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/973620295bd444fca5b90aaef0156f99 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59f3a94b66124377a51ccb72a72a32f0 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c9cd231ca6c41e691fb72d8be1e56e5 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/353a4bc2d8494764a1c28fa0646ccf11 : SUCCESS in 20m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/82c7c578422f48fbbcf6ae00ee6f7535 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b7acb1d3645d4a8cbb2bbbe7ca198e59 : SUCCESS in 59m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4aec250d16cd4073bfa75ec5521f2e0d : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/56a72b96ffe5488c8d4b75157e635f52 : SUCCESS in 1h 28m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/32598b09e15744ef982c472028b684f7 : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bf28d8a7a6e94d7f9ea2b491a313f85f : SUCCESS in 28m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9e2a432d80c44598a36906da8be15269 : SUCCESS in 12m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"ef4d34e7b51ca3d58b545f1418f0623a9b215885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:23:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7f7264eed747ec86a92f4da1e1fb44 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e135db7308f476f8940aec141b18435 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"64fd8e17263a65275d9acdc5792193890face82f","revisions":{"70842c74e13c6204211e2dc8336869388d7f7991":{"kind":"REWORK","_number":1,"created":"2020-09-29 14:01:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/1"}}},"commit":{"parents":[{"commit":"4325baa61bda9b5f9378944870d863695da7d9c5","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4325baa61bda9b5f9378944870d863695da7d9c5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"subject":"WIP: Disable overcloud networks for Ironic when unused","message":"WIP: Disable overcloud networks for Ironic when unused\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70842c74e13c6204211e2dc8336869388d7f7991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70842c74e13c6204211e2dc8336869388d7f7991"}]},"branch":"refs/heads/master"},"470b0d38b4b1e0df8e29b387f95cb643f0a87b81":{"kind":"REWORK","_number":2,"created":"2020-09-29 16:22:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/2"}}},"commit":{"parents":[{"commit":"4325baa61bda9b5f9378944870d863695da7d9c5","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4325baa61bda9b5f9378944870d863695da7d9c5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 16:22:28.000000000","tz":60},"subject":"WIP: Disable overcloud networks for Ironic when unused","message":"WIP: Disable overcloud networks for Ironic when unused\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/470b0d38b4b1e0df8e29b387f95cb643f0a87b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/470b0d38b4b1e0df8e29b387f95cb643f0a87b81"}]},"branch":"refs/heads/master"},"290b33cc6a3918ffe526bd4351ca9aff5c27b0c6":{"kind":"REWORK","_number":3,"created":"2020-09-29 18:16:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/3"}}},"commit":{"parents":[{"commit":"4325baa61bda9b5f9378944870d863695da7d9c5","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4325baa61bda9b5f9378944870d863695da7d9c5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 18:16:00.000000000","tz":60},"subject":"WIP: Disable overcloud networks for Ironic when unused","message":"WIP: Disable overcloud networks for Ironic when unused\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/290b33cc6a3918ffe526bd4351ca9aff5c27b0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/290b33cc6a3918ffe526bd4351ca9aff5c27b0c6"}]},"branch":"refs/heads/master"},"eac80159c745e578c8f5c22e5119c8b60fba6875":{"kind":"REWORK","_number":4,"created":"2020-09-30 08:09:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/4"}}},"commit":{"parents":[{"commit":"4325baa61bda9b5f9378944870d863695da7d9c5","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4325baa61bda9b5f9378944870d863695da7d9c5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-30 08:09:23.000000000","tz":0},"subject":"WIP: Disable overcloud networks for Ironic when unused","message":"WIP: Disable overcloud networks for Ironic when unused\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eac80159c745e578c8f5c22e5119c8b60fba6875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eac80159c745e578c8f5c22e5119c8b60fba6875"}]},"branch":"refs/heads/master"},"33f360be4f71d41c94b62b9b30c65f6cbee039de":{"kind":"REWORK","_number":5,"created":"2020-10-01 10:17:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/5"}}},"commit":{"parents":[{"commit":"d0385b84eb61f53bc710f6f276a0d921f4cfaf91","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0385b84eb61f53bc710f6f276a0d921f4cfaf91"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:10:28.000000000","tz":60},"subject":"Disable overcloud networks for Ironic when unused","message":"Disable overcloud networks for Ironic when unused\n\nWe do not need the workload provisioning, cleaning or inspection\nnetworks when Ironic is disabled in the overcloud.\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f360be4f71d41c94b62b9b30c65f6cbee039de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f360be4f71d41c94b62b9b30c65f6cbee039de"}]},"branch":"refs/heads/master"},"e4ee5953de38256d8283d93a929e324d98070d2e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-02 10:03:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/755015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/6"}}},"commit":{"parents":[{"commit":"f9de6a025d4b986ed9461c67485f26fc953995c9","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9de6a025d4b986ed9461c67485f26fc953995c9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-02 10:03:18.000000000","tz":0},"subject":"Disable overcloud networks for Ironic when unused","message":"Disable overcloud networks for Ironic when unused\n\nWe do not need the workload provisioning, cleaning or inspection\nnetworks when Ironic is disabled in the overcloud.\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4ee5953de38256d8283d93a929e324d98070d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4ee5953de38256d8283d93a929e324d98070d2e"}]},"branch":"refs/heads/master"},"64fd8e17263a65275d9acdc5792193890face82f":{"kind":"REWORK","_number":7,"created":"2020-10-02 14:23:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/755015/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/755015/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/755015/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/755015/7"}}},"commit":{"parents":[{"commit":"f9de6a025d4b986ed9461c67485f26fc953995c9","subject":"Disable ironic in the overcloud by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9de6a025d4b986ed9461c67485f26fc953995c9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-29 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-02 14:23:10.000000000","tz":0},"subject":"Disable overcloud networks for Ironic when unused","message":"Disable overcloud networks for Ironic when unused\n\nWe do not need the workload provisioning, cleaning or inspection\nnetworks when Ironic is disabled in the overcloud.\n\nChange-Id: I300d0ef136224126f25d2c70a80a42afeea5f586\nStory: 2008207\nTask: 40992\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64fd8e17263a65275d9acdc5792193890face82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64fd8e17263a65275d9acdc5792193890face82f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
