)]}'
{"id":"openstack%2Fkayobe~755342","triplet_id":"openstack%2Fkayobe~master~I55af5d979bdbee9dd963845204c2f082fa7b27d7","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I55af5d979bdbee9dd963845204c2f082fa7b27d7","subject":"CI: workaround overcloud upgrade issue with ironic","status":"MERGED","created":"2020-09-30 16:20:03.000000000","updated":"2020-10-01 07:52:47.000000000","submitted":"2020-09-30 21:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755342-1601500151141-bdf3d015","meta_rev_id":"be0a356feccf082f9d082adf0d8fedb13961e590","_number":755342,"virtual_id_number":755342,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 21:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-30 19:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-30 19:05:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 21:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb6694d169b2f0fc17b8855f3adb548519b6aa04","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-30 16:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f14edd117c2926e94380acdca2addcd272eaf91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:07:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a909554a66b4f2da3e7eb712b75ec94 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97709d1911e74245bd2984ab266abf7b : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af79d9056bf444099c00e5a3d41de8f2 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa32b4ef94544993859dc9dc55974b12 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9b13efb2ee46b8b0272e51545f0d88 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ecd77b813d849bc83faa2c02393f775 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b7dd4d1ca9d4bd8a79905c7fd07da38 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fc17c3b0f024dbca0ed27405b07518f : SUCCESS in 21m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a1065f37a16e4c8da3eb0fabaaaa761c : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0859f3a63790444a9cdfba1ad7fd7e83 : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/38e99e378bc844b9b1eade253622b63d : SUCCESS in 1h 43m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/682e5a3d23904352ad532d48b4eabef4 : SUCCESS in 16m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7384b591c6854e13a692834804d47378 : SUCCESS in 28m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c4d77eddbe944949953d709b03fe042e : SUCCESS in 12m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e1ee443860b424a004cf72a8d0a30406504d0c0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-30 19:05:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bab404c62ea750b5fcc5ec6c6e2496b6ea3b4c61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:06:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cd1ef7ced5d066c14ab950269312be1a5cae048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:09:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d035139efa46dab340b1235d51dca1 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01da30097f304fd09c9ae4f918d69b27 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8180f901b5a1487089cd6ba6d7626abe : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d9ac6251d24dbca2f3cfffeaa20b2f : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01b85a279d574c40a2ee3da8d432b326 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eec7cc469a7f4e2483b39f74ea1d2608 : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3477c47f9b5d44c78c5b37bd7151958c : SUCCESS in 20m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9fd97d0c091c481eb166472860d50c89 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/47ca02b268694f0b8821d0abe5f10b52 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0c82ce60b9fc4723b7546a5de53f295c : SUCCESS in 1h 57m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab54b07d96cd48a5a61e1af5b06faee2 : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7576ee8317d24da0b9aa1d8933a6bcc4 : SUCCESS in 26m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/833384e80ac84835a88313b185cd9a45 : SUCCESS in 9m 50s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"df31875e5751498db20e412e4baa9c3b7ca7ad45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:09:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b24ef9c48770de521b1348d2ff36d58bd45b98f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:10:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbbce084f4474588a19ca101e2a79423 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"be0a356feccf082f9d082adf0d8fedb13961e590","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 07:52:47.000000000","message":"Created a revert of this change as Icd7b0f3b02bf3f22468daa8893c043ac9f3be4d6","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1","revisions":{"bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1":{"kind":"REWORK","_number":1,"created":"2020-09-30 16:20:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/755342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/755342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/755342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/755342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/755342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/755342/1"}}},"commit":{"parents":[{"commit":"7c25fec6ca9526e7c7f12c05d45feb6063a88974","subject":"Merge \"Remove unneeded doc requirements in molecule tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c25fec6ca9526e7c7f12c05d45feb6063a88974"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-30 16:18:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-30 16:19:34.000000000","tz":60},"subject":"CI: workaround overcloud upgrade issue with ironic","message":"CI: workaround overcloud upgrade issue with ironic\n\nIronic has an issue where nodes will be switched to use the direct\ndeploy interface during upgrade, if it is enabled. This should be fixed\nvia https://review.opendev.org/755295, after which this patch should be\nreverted.\n\nChange-Id: I55af5d979bdbee9dd963845204c2f082fa7b27d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb9a595e5ea16b5ad6500fc57cbee3f1d4035ab1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
