)]}'
{"id":"openstack%2Fkayobe~755474","triplet_id":"openstack%2Fkayobe~master~Id3128380f5ff74d24265f6b2132c6d7992bf00ba","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id3128380f5ff74d24265f6b2132c6d7992bf00ba","subject":"dev: fix test scripts when ironic is disabled","status":"MERGED","created":"2020-10-01 10:17:00.000000000","updated":"2020-10-06 14:28:06.000000000","submitted":"2020-10-06 14:26:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755474-1601994389468-89f32885","meta_rev_id":"cef85c802b1a92700782a6ade2b01b5f8b929986","_number":755474,"virtual_id_number":755474,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 14:26:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-10-06 12:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-10-06 11:25:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 19:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-10-06 12:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 19:58:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-06 11:25:26.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2020-10-06 12:22:29.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-10-06 14:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e641755a275c896477edc457040e6cf8ca9630b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 10:17:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc5f7771edb5035e525a0cd41f7d3b2e654ddf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f300b195f614c57820be92c44be1ba4 : SUCCESS in 41m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dc120c48b847456b8e6c3802cca7a332 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d9b03a06790ee13319de6e4f584be00225e6ade6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 12:39:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"87fb440b2d7dd42454f78664326f8b7ccaf5cca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 15:02:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66715b9be42d44bfa17650af017ae6f2 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f470e379f61d4b528b371fc9eccdd092 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b502b951e7d488f8a1038f22df6138f : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8acff5b1153f46948db5b6e108c27923 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f573ad83d96439ebb7c7c34ee7f2886 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d4c4456942d4da69684fa14f602d60e : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94939921149c485db7f9e8afdb0bf794 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e370facc856444bab7c2f5345980573e : SUCCESS in 23m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cdfa2e93bb204a85a3b6bfcce4def22c : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1e0da5b47c4e4f0382bdb592b6858305 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/38452e5ff6e84640a292ee9475937c11 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/daf3c69f0e034d69a4b904540160cb95 : SUCCESS in 1h 51m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7643323304bd4cb49a6a972d7380abb2 : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2c513dd957443e29964164eb2b0d42e : SUCCESS in 29m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7a5530f327eb483ba0ec5588a1b753b3 : SUCCESS in 12m 32s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"5b2840428f093354acc686ce925e6417c994e643","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 10:03:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2be6b0f2e192214200891c7104f7554f1fabf45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 12:20:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1be8ae0f0cd48eab3f01bfab2ce2912 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9be4839ddeb4c0481b9db35a8667952 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abfc0e541c494ac787a5ae2aee88f9d6 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9adb47fc8f94ada8473378454900d43 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068d2e6fee4b43f6a504aa419223dab6 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/497e679a31024286b239015466e69b62 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a27db8265a5140389f02585390f4dd9d : SUCCESS in 9m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5474c4eba67f48e5860b744f9359417c : SUCCESS in 16m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/734afd8011a7492bb3afd0d9fdb0c65d : SUCCESS in 1h 03m 22s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ae7dafb916d8485cb8e66f18c3d0a6dc : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d344a4d225c54b529675f7db96004159 : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/32796a85bd26471d98fff700b2303b9a : SUCCESS in 2h 01m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8fcd0b50e3f848459614950f54bc4b44 : SUCCESS in 15m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9ad2da9208f544a3bde5448378fcfba1 : SUCCESS in 29m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5d08cdcfb3d442ac8e10280e9eb6cdf1 : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"cc29990e3a34871e417a510eb35b3f79c6f5a4e4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-02 13:45:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70697790cf2b8df215e158ceef5d7b034f8766e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:02:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"462ec640178cce58d36d6f904553e302c8eb9598","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:24:30.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bacb527d569ffc08e1272e2600df0af2f7822357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 16:30:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ba9d3c28de4cc0a36b162a3905b6b2 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be3eeeb2e71f48a0add1f86590a545e3 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cf7753f15d44cd7a50d2b14fe9af1b0 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b7441def07471db135478873d14139 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c54b01cbbf54fd8a7d76eb1563770bb : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ab45444f0ac4722b4600df6cfdd7b60 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ced4b5c3de414dd887575a888fb99e1c : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1a96c0b503e4fb28b58b8cc0ea3ff5c : SUCCESS in 19m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/099a619d42514d479213bb6e333d1da7 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2f80135a98ea4fc58e88cecefab5f11e : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d244368de9a84969abb53ba469b816ae : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/20495cb63b4d4fa69f3b3a85356d4ff5 : SUCCESS in 1h 59m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/40b5ff193af144d0a1fb88333104fadf : SUCCESS in 19m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a7dce2c8ecdc42fdaf76d940d3157749 : SUCCESS in 26m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/087070fc2911445db506f938562fe475 : SUCCESS in 12m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"88b90d67741d9c8b61b00df0c0f54a55a56708fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-05 19:58:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0f8c1cfd86553ca7cdea6bbd91aa72223af4ba44","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-10-06 11:25:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"089bdd312610c0012644a4e703834b0cf06b8fe7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-10-06 12:22:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"38cf9b1faaef9cd07d16593510d6d63f813908c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 12:22:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f42d6088fb2b8b69190628b6e52df485ccef3a4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:26:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"acc5f9d752ecc218d31c76de3914471e7ac4d1ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:26:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd783754748a4bb2bdf2d9ef627b149d : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9869d354ed5a4c74ba208e18b065de0a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9451200144e44818b69d9ac9bdbf6505 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1f425d91784fb7af5f30c29be25c20 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01298824934a4b8ca863d4bc6f782532 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5a6e803ca0e4625a40af042c21991ca : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40a7ecaabcc04d029596ce29c1e5a1c9 : SUCCESS in 20m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/01d9c16c5bbf4cd2a09de368feb088af : SUCCESS in 1h 10m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1b13d37c26a9431cbd4394a850d8715c : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/02fb3456a0094d2aa5a40418ddfe1fcf : SUCCESS in 17m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1b4fb17a364c4eab9c441d744c74979b : SUCCESS in 1h 56m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4bb423bc6bbb4670bfe9b97fb9616d57 : SUCCESS in 19m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aaf32ccfd7e14a2b876110f0135ba60e : SUCCESS in 25m 12s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fa443a3641964354b9e384e7f9197aaa : SUCCESS in 9m 56s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"cef85c802b1a92700782a6ade2b01b5f8b929986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:28:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b65f779ffb443cb889d8f368956a4d : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6a4e7c4e91a800704b6b42c5b3d7e8921c1ccbbb","revisions":{"e17050281410507c0cb92b371819625f1900272d":{"kind":"REWORK","_number":1,"created":"2020-10-01 10:17:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/755474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/755474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/755474/1"}}},"commit":{"parents":[{"commit":"33f360be4f71d41c94b62b9b30c65f6cbee039de","subject":"Disable overcloud networks for Ironic when unused","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f360be4f71d41c94b62b9b30c65f6cbee039de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:12:26.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:15:52.000000000","tz":60},"subject":"dev: fix test scripts when ironic is disabled","message":"dev: fix test scripts when ironic is disabled\n\nWhile we always test baremetal compute in CI, development environments\nmay not. Given that Ironic is now disabled by default, we should make\nthis work out of the box.\n\nStory: 2008207\nTask: 41003\n\nChange-Id: Id3128380f5ff74d24265f6b2132c6d7992bf00ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e17050281410507c0cb92b371819625f1900272d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e17050281410507c0cb92b371819625f1900272d"}]},"branch":"refs/heads/master"},"14deb4720a710a0d90bcd3053747dc80660c2a73":{"kind":"REWORK","_number":2,"created":"2020-10-01 12:39:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/755474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/755474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/755474/2"}}},"commit":{"parents":[{"commit":"33f360be4f71d41c94b62b9b30c65f6cbee039de","subject":"Disable overcloud networks for Ironic when unused","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33f360be4f71d41c94b62b9b30c65f6cbee039de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:12:26.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 12:39:11.000000000","tz":0},"subject":"dev: fix test scripts when ironic is disabled","message":"dev: fix test scripts when ironic is disabled\n\nWhile we always test baremetal compute in CI, development environments\nmay not. Given that Ironic is now disabled by default, we should make\nthis work out of the box.\n\nStory: 2008207\nTask: 41003\n\nChange-Id: Id3128380f5ff74d24265f6b2132c6d7992bf00ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14deb4720a710a0d90bcd3053747dc80660c2a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14deb4720a710a0d90bcd3053747dc80660c2a73"}]},"branch":"refs/heads/master"},"99618954a56bf31d5933f18029b955bc49032faf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-02 10:03:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/755474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/755474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/755474/3"}}},"commit":{"parents":[{"commit":"e4ee5953de38256d8283d93a929e324d98070d2e","subject":"Disable overcloud networks for Ironic when unused","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4ee5953de38256d8283d93a929e324d98070d2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:12:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-10-02 10:03:50.000000000","tz":0},"subject":"dev: fix test scripts when ironic is disabled","message":"dev: fix test scripts when ironic is disabled\n\nWhile we always test baremetal compute in CI, development environments\nmay not. Given that Ironic is now disabled by default, we should make\nthis work out of the box.\n\nStory: 2008207\nTask: 41003\n\nChange-Id: Id3128380f5ff74d24265f6b2132c6d7992bf00ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99618954a56bf31d5933f18029b955bc49032faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99618954a56bf31d5933f18029b955bc49032faf"}]},"branch":"refs/heads/master"},"6a4e7c4e91a800704b6b42c5b3d7e8921c1ccbbb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-10-02 14:24:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/755474/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/755474/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/755474/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/755474/4"}}},"commit":{"parents":[{"commit":"64fd8e17263a65275d9acdc5792193890face82f","subject":"Disable overcloud networks for Ironic when unused","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64fd8e17263a65275d9acdc5792193890face82f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-01 10:12:26.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-02 14:24:30.000000000","tz":0},"subject":"dev: fix test scripts when ironic is disabled","message":"dev: fix test scripts when ironic is disabled\n\nWhile we always test baremetal compute in CI, development environments\nmay not. Given that Ironic is now disabled by default, we should make\nthis work out of the box.\n\nStory: 2008207\nTask: 41003\n\nChange-Id: Id3128380f5ff74d24265f6b2132c6d7992bf00ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a4e7c4e91a800704b6b42c5b3d7e8921c1ccbbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a4e7c4e91a800704b6b42c5b3d7e8921c1ccbbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
