)]}'
{"id":"openstack%2Fkayobe~755804","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ib417837d0772338b16ea9f7f2540549f277d5de8","project":"openstack/kayobe","branch":"stable/ussuri","topic":"bug/epel-disable-stable/ussuri","hashtags":[],"change_id":"Ib417837d0772338b16ea9f7f2540549f277d5de8","subject":"When EPEL install flag is not set","status":"MERGED","created":"2020-10-02 14:26:13.000000000","updated":"2020-10-05 18:53:30.000000000","submitted":"2020-10-05 18:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755804-1601923650869-72bff4cb","meta_rev_id":"7bbbb97aba5159c51d01d3002ec8c15f7264b744","_number":755804,"virtual_id_number":755804,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-05 18:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-10-03 14:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-05 16:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-05 16:41:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"date":"2020-10-02 14:26:13.000000000","_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 14:26:13.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-10-05 16:41:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-10-05 18:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"430f564dbbd694f629cade7be9ccedf1b62d2252","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-02 14:26:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02f521d9f43e5b6a4c36b5598db0a6fbbffc94a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 17:00:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c18211541f44d409f48d4d299629b7a : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f16ec529de4a42b5983b402676de414f : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d65ee4f810294a84ac57eb77a7550654 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dffe293b07c496cb4415bd5d9591faa : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650c589f12b64c3598b00f986dfaa7f1 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3750150327247d6a1735de30f203a32 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bfdf4e814d0d4f049fdb6cf89595e1f0 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5834dacc14a4405b92b4d02e6b4c25e9 : SUCCESS in 16m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8fd2613dca1e454fbf97a4a18a294260 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8ff047b510ec4864999ef4875e1ceaec : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f92f8080a57f4faa960bb4c48c24969d : SUCCESS in 2h 24m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a731761546494fbe98c719419d50e13f : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9c15115fc5ba48afaab1e3fea1f81db9 : SUCCESS in 43m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ce14091b705147fdacb520bc8e7a9b75 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ff58c5d9acc0a8ef158e7077d3f03898a0676f74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-03 14:59:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a241d452057edbcdda903313ff789d3f6b5a779","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-10-05 16:41:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fd2eb35917241056b42c472d1f73a84973d70c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:41:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"771292b2b989f45eb6b2f5acbaaaff0be4cffc59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 18:47:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e429c479ec4a1ab9298048a651f8f9 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/433c4df5bc364ad384f76cc4d98930f5 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccff874bd3c54885a0362dfd7e9f747e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139846d7206644929e128d4df7dd9a04 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4040093d60d45b5a0d0368cfa44647b : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f499f407af564a0a8669196fbf642fb2 : SUCCESS in 7m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c5fff0eb3764659bf8c90733c1a7a0f : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e33bc79d55be4bd3a8ec8a0c48678521 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3c77f6972a4f4ba9a10bb2e934032fed : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/845907c6bbf84ce1b20f01026a7f25fd : SUCCESS in 1h 47m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f92fc7d189884b0eb262066f32794c3b : SUCCESS in 17m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/707c4a02b97945049293ce91d09b20ce : SUCCESS in 26m 42s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/62901a15585f44c6813b6ce02e7972d3 : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c57d73f5e73ce419071feca15fdbb090c430f657","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 18:47:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b5d6323961f4baa9775da4b0738f214407ff8db8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 18:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe6c012ebb94dd4b2553de726622b64 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b1ca2485c56428de461b480326db6a5dde8f2e1","revisions":{"6b1ca2485c56428de461b480326db6a5dde8f2e1":{"kind":"REWORK","_number":1,"created":"2020-10-02 14:26:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/755804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/755804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/755804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/755804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/755804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/755804/1"}}},"commit":{"parents":[{"commit":"eb063dbcd957e9f1b36724081815f402923e2ef8","subject":"Performance: avoid set_fact in Kolla Ansible host vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb063dbcd957e9f1b36724081815f402923e2ef8"}]}],"author":{"name":"Stig Telfer","email":"stig.openstack@telfer.org","date":"2020-09-30 16:49:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-10-02 14:26:13.000000000","tz":0},"subject":"When EPEL install flag is not set","message":"When EPEL install flag is not set\n\nThe flag dnf_install_epel was being used to predicate installation\nof the EPEL RPM repo package, but not preventing the installation\nof configuration for access to EPEL.  This small patch completes\nthe job.\n\nUpdates the overcloud host configure CI job to install EPEL mirrors,\nsince it was previously relying on this buggy behaviour.\n\nChange-Id: Ib417837d0772338b16ea9f7f2540549f277d5de8\n(cherry picked from commit e5a2818896a47c902487ea9aa97b5a4a03f2b9d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b1ca2485c56428de461b480326db6a5dde8f2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b1ca2485c56428de461b480326db6a5dde8f2e1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
